)]}'
{"id":"openstack%2Fskyline-console~915254","triplet_id":"openstack%2Fskyline-console~master~I34e98bb6950105a7bb89a7671fcda93564349ad5","project":"openstack/skyline-console","branch":"master","topic":"styles","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-04-09 04:42:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34e98bb6950105a7bb89a7671fcda93564349ad5","subject":"feat: optimize menu display","status":"MERGED","created":"2024-04-08 08:44:57.000000000","updated":"2024-04-09 04:43:53.000000000","submitted":"2024-04-09 04:42:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915254-styles","meta_rev_id":"3b6ce5a7cc5d36c888cd109de4e9e0592731ac24","_number":915254,"virtual_id_number":915254,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-09 04:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-04-09 04:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-04-09 04:30:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-08 08:58:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-09 04:30:42.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"2b3a43a5c472b52c7f2e9f3d14d4c55fa1069e47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-08 08:44:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e159f44f74c1264f20d3a7081c64a4a847af18b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 08:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b92bad06fb4e8f8baa273708d04972\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a4b60f5a48541d5aec26e13035f9642 : SUCCESS in 4m 26s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7774488c88db41fda785aba8a5d917ba : SUCCESS in 7m 55s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1d51acb6ce8b4d37af80a10662344de4 : SUCCESS in 7m 57s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/03bf261870b143b9834a6e952e7976d9 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3604ad8030618c45d26cf41dc227bd98f7cd0562","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-04-09 04:30:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d29041ba7e3fae92b1882d2c9ee425e380ce775a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 04:31:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06a932ef12da669f3c6759a0de157802718d2c4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 04:42:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8f1899dc1a84e45b722f2ad07a2049f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db70e09dc3847b99ab7cb51e7c47d9f : SUCCESS in 5m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f6e82b62937f46b490e4ab0897b7232a : SUCCESS in 6m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/81013a04d20945338a871bb126a8a2d1 : SUCCESS in 7m 02s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d7d86de3202e49b8bc3707697cbe92b3 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a4d388a8248d4d7361feea99abe845a077e4c277","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 04:42:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3b6ce5a7cc5d36c888cd109de4e9e0592731ac24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 04:43:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8608afb9ebe04298a36e47aff25d51f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9cddb555c08412295e37bb44a61fd84 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"791176a4855b9e407426c4039b5d3c72081767f0","revisions":{"791176a4855b9e407426c4039b5d3c72081767f0":{"kind":"REWORK","_number":1,"created":"2024-04-08 08:44:57.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/54/915254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/54/915254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/54/915254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/54/915254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/54/915254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/54/915254/1"}}},"commit":{"parents":[{"commit":"2c05f14a5416113161fa5420728a8224ca93f0d8","subject":"Merge \"feat: update code format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2c05f14a5416113161fa5420728a8224ca93f0d8"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-08 08:42:51.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-08 08:44:36.000000000","tz":480},"subject":"feat: optimize menu display","message":"feat: optimize menu display\n\nAdjust the maximum number of characters in the menu based on the current language\n\nChange-Id: I34e98bb6950105a7bb89a7671fcda93564349ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/791176a4855b9e407426c4039b5d3c72081767f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/791176a4855b9e407426c4039b5d3c72081767f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
