)]}'
{"id":"openstack%2Fskyline-console~915800","triplet_id":"openstack%2Fskyline-console~master~Ib17f3c84aade6ce44610f379e577e676c30facb7","project":"openstack/skyline-console","branch":"master","topic":"port","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-04-15 07:53:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib17f3c84aade6ce44610f379e577e676c30facb7","subject":"feat: update qos policy info","status":"MERGED","created":"2024-04-15 05:13:08.000000000","updated":"2024-04-15 07:54:58.000000000","submitted":"2024-04-15 07:53:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915800-port","meta_rev_id":"c30136afff122bf48abb9a13fa5e46a9f3959e2c","_number":915800,"virtual_id_number":915800,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-15 07:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-04-15 07:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-04-15 07:03:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 06:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-15 07:03:57.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"e178494372bab04a4da56c22ea5c40e5fd929f7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-15 05:13:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea4e331806588c8cfcbedd97710cdc96d2c825e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 06:04:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddcf3c2d27a649489a5cb9d6af350487\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6589625208af4499b0f0de0df01f279a : SUCCESS in 4m 00s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f43acfde661047b8ac8a31a3dd2a7d7d : SUCCESS in 10m 53s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/940958541e314614adbca7946ef7a57b : SUCCESS in 6m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8b9cb50f3f234d23a9c55b771f74ef09 : SUCCESS in 6m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5deb730b546e443683b4e29d5d89b0c5 : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5d6e6d4190db2e776e515f9b64ad1007fa383e5a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-04-15 07:03:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2d7d2e06d543a6533320cfc8e68f4107339cc49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 07:04:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e93a8c57cfefe1d95a8fba1b75afbafd69ea5744","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 07:53:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/717091aac02d4f7e8925b5e38bd893bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc31cc844534acb87896a7d08595a24 : SUCCESS in 4m 02s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/67105fd6ccfc4159b1ebbcdd98351ca1 : SUCCESS in 6m 11s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0e78df295e334ea8924523153feb0d78 : SUCCESS in 4m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/cfb5dc02d74345e3ae9459fc74970504 : SUCCESS in 4m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a4729f8c6af147e7aa91aa04ee22a34f : SUCCESS in 43m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bf702651dee4405f6cdf0051c26f6b666fe4656f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 07:53:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c30136afff122bf48abb9a13fa5e46a9f3959e2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 07:54:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/394011e08a024b83934e620d48494502\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c7a84723314b89862faec7d2cdd319 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34a41dd38dc22dafa32c1c6fd0162d2a4785ca80","revisions":{"34a41dd38dc22dafa32c1c6fd0162d2a4785ca80":{"kind":"REWORK","_number":1,"created":"2024-04-15 05:13:08.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/00/915800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/00/915800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/915800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/915800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/915800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/00/915800/1"}}},"commit":{"parents":[{"commit":"84d1e7af720e2dc71f02e59490d92a66c4dcb7aa","subject":"fix: update long title style for menu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/84d1e7af720e2dc71f02e59490d92a66c4dcb7aa"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-15 05:11:34.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-15 05:11:43.000000000","tz":480},"subject":"feat: update qos policy info","message":"feat: update qos policy info\n\nFor the qos policy info in the port detail page:\n1. add qos policy name\n2. update qos policy link for console and administrator\n\nChange-Id: Ib17f3c84aade6ce44610f379e577e676c30facb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/34a41dd38dc22dafa32c1c6fd0162d2a4785ca80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/34a41dd38dc22dafa32c1c6fd0162d2a4785ca80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
