)]}'
{"id":"openstack%2Fskyline-console~916030","triplet_id":"openstack%2Fskyline-console~master~I647a7c69da6d463c11fe98ba5e33bf405ae14e9d","project":"openstack/skyline-console","branch":"master","topic":"identity","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-04-17 05:20:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I647a7c69da6d463c11fe98ba5e33bf405ae14e9d","subject":"feat: update user group role check","status":"MERGED","created":"2024-04-17 02:19:57.000000000","updated":"2024-04-17 05:21:49.000000000","submitted":"2024-04-17 05:20:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916030-identity","meta_rev_id":"cf9fbdcd9ffffee7a981e376a5cbe285fd6c0446","_number":916030,"virtual_id_number":916030,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 05:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-04-17 04:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-04-17 04:40:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 03:34:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 04:40:38.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a30b95d17890cf7faed79efd5dfa1643f4834e88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-17 02:19:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60eb4585322e5fa700cc4bb046c6e13d7c62adec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 03:34:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d18589b557454a81600cee40bcbb40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05ff3cc5f074a52bdfc9bac5a8ec505 : SUCCESS in 4m 22s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/73e8f8df557f4ae4a2ea2c335c91a09e : SUCCESS in 10m 08s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c57e89dfdc494a6d856d01473261d913 : SUCCESS in 6m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3b6f90af5293472086d3878f316cbc3c : SUCCESS in 8m 46s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f7086bda11e944e18fb9f134eb063c4a : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5b6f9b286fb347314e7f209202581cc680c3da56","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-04-17 04:40:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84ca8170e05c9b33e921856d4f5aa81275460f03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 04:41:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5597b30a2061066e14f524da47b3bfa1076161b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 05:20:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9008bb95227b4f78914a7c391464c28d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df899dacdfb647f7a84722527526563a : SUCCESS in 4m 08s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/261b4f4cafc94899ae580a6091434010 : SUCCESS in 6m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e6447793ca164ed38085e92f8d5d754c : SUCCESS in 7m 21s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/323a45751e144a49b1054338f3be7930 : SUCCESS in 4m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8349a3a49b8942a99f9fa9b9f05a2cd3 : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":1},{"id":"689a7d7a4574f4f8a071b4991e656badbc5b3a2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 05:20:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf9fbdcd9ffffee7a981e376a5cbe285fd6c0446","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 05:21:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d00cc04ee13042229a0dc5c835b54165\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f959d04fc9324a3abce04350b7c2f257 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0be5a1294a280a31ff25a3c3dd5989c34cfa535","revisions":{"a0be5a1294a280a31ff25a3c3dd5989c34cfa535":{"kind":"REWORK","_number":1,"created":"2024-04-17 02:19:57.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/30/916030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/30/916030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/916030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/916030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/916030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/30/916030/1"}}},"commit":{"parents":[{"commit":"3c71281280c642e31590f2e862346aacb2c2e0fc","subject":"feat: update subnet actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3c71281280c642e31590f2e862346aacb2c2e0fc"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-17 02:16:38.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-17 02:19:16.000000000","tz":480},"subject":"feat: update user group role check","message":"feat: update user group role check\n\nWhen adding a user group to a project, at least one role must be selected.\n\nChange-Id: I647a7c69da6d463c11fe98ba5e33bf405ae14e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a0be5a1294a280a31ff25a3c3dd5989c34cfa535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a0be5a1294a280a31ff25a3c3dd5989c34cfa535"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
