)]}'
{"id":"openstack%2Fskyline-console~916031","triplet_id":"openstack%2Fskyline-console~master~I619ec9ada74c5040bdb234f375651c8b053db79b","project":"openstack/skyline-console","branch":"master","topic":"heat","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-04-17 05:41:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I619ec9ada74c5040bdb234f375651c8b053db79b","subject":"feat: update creating heat stack","status":"MERGED","created":"2024-04-17 03:09:17.000000000","updated":"2024-04-17 05:43:14.000000000","submitted":"2024-04-17 05:41:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916031-heat","meta_rev_id":"d5b796afd7afa60d033936d64fe18c5ce9bde3f0","_number":916031,"virtual_id_number":916031,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 05:41:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-04-17 04:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-04-17 04:41:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 03:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 04:41:00.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"459e8120b32d3593717159628e87c4c74fdeec26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-17 03:09:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94b063327fd3bd6a22552a45a1ff63f4c7639286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 03:55:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4261a762f0b6491da8ccc05a0ffa994e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be83f4528b34b7198f46e01a23c9b0e : SUCCESS in 4m 35s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fdecca23eeda4ab9bec57bad9283cf49 : SUCCESS in 10m 53s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/39e22ee9be0f4c708b5e2fd6b7ee1dc2 : SUCCESS in 8m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a1ce09e05dae4d3a944ef779a6cbc9e5 : SUCCESS in 8m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/339940f5a6d34a0ebd6fb1d42dc8f3d8 : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6d8f53fbd67e59724cb0a40d89339a21805a73c5","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-04-17 04:41:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92faccd267242e761f556bc1ee41dbbba8b10206","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 04:41:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c7282b776988a22b89706247bd9c834d0d1907f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 05:41:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a02b4c13a34bb2b0ae0ae1dcf6c1ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ded34291a843ef89bf6b7d6b5e2f0e : SUCCESS in 4m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8499d7f73e9c469bbff850abdf0a4acd : SUCCESS in 10m 54s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c6c59bd1e01e4d6a952ca87caba18bb1 : SUCCESS in 5m 01s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4760d8a46f554bee8439ca83b83544ec : SUCCESS in 6m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/53c26d26fdd04eef9ece6e9bc153b543 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":1},{"id":"15f98bd3241da268ab4a5fbe259624600fca2824","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 05:41:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d5b796afd7afa60d033936d64fe18c5ce9bde3f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 05:43:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb5c799b33e24b3eae282de380d5cf5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a0fa9aaadb4f0a87bdd8530ee936fc : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"952543f5aedd07ddd5f1fe7d985c65267a785db7","revisions":{"952543f5aedd07ddd5f1fe7d985c65267a785db7":{"kind":"REWORK","_number":1,"created":"2024-04-17 03:09:17.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/31/916031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/31/916031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/916031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/916031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/916031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/31/916031/1"}}},"commit":{"parents":[{"commit":"3c71281280c642e31590f2e862346aacb2c2e0fc","subject":"feat: update subnet actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3c71281280c642e31590f2e862346aacb2c2e0fc"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-17 03:08:47.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-17 03:08:47.000000000","tz":480},"subject":"feat: update creating heat stack","message":"feat: update creating heat stack\n\nWhen the parameters in the second step are changed, the contents of the environment variables in the first step are updated simultaneously.\n\nChange-Id: I619ec9ada74c5040bdb234f375651c8b053db79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/952543f5aedd07ddd5f1fe7d985c65267a785db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/952543f5aedd07ddd5f1fe7d985c65267a785db7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
