)]}'
{"id":"openstack%2Fskyline-console~916153","triplet_id":"openstack%2Fskyline-console~master~I067120da290937281fcd1daf2726d6c398251b24","project":"openstack/skyline-console","branch":"master","topic":"security-group","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-04-18 07:49:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I067120da290937281fcd1daf2726d6c398251b24","subject":"fix: update ipv6-encap i18n","status":"MERGED","created":"2024-04-18 02:19:49.000000000","updated":"2024-04-18 09:38:02.000000000","submitted":"2024-04-18 09:37:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916153-security-group","meta_rev_id":"103cc20c690c5aa69461822c83b0b2daaad1582e","_number":916153,"virtual_id_number":916153,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-18 09:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-04-18 07:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-04-18 07:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-18 02:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-18 07:19:09.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"9e48d1f31f213b4c48e958f323b74d0c731f3469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-18 02:19:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69fe758b7f4976a7a2ed5cb37dec3cb07aa4c862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 02:51:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aca50a6b9cb842428468e3c69e9792d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a8bc6ad253453ca95446edb0676394 : SUCCESS in 5m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/38907ae3cb5c4a9e84d8daff0759e92b : SUCCESS in 9m 52s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/209ab704c8bb4a5f9264044b4ad511c8 : SUCCESS in 7m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c30191f5e1014968b80fb14b32389a01 : SUCCESS in 6m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c6082ee9ac074c1e90fa55e870471845 : FAILURE in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"52bbbad659e55857adfe99ce719dbba01c98593f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-18 05:55:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4d3bed3535fbc727273593a9b9c6e728429ba31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 07:05:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec9e5d00379a49a3b8dc9db9935b2f6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e006a5761442c496ed558a51626eab : SUCCESS in 4m 42s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6b7ab5fe61a04dd09f5dea61fb3b4ca5 : SUCCESS in 10m 25s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/39dc8cdae1cb4d8ab651d94c46e6498f : SUCCESS in 4m 23s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/974e9b8b9a3d4b77a16aa673743fae8d : SUCCESS in 4m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bb0c6bff4fcd4e05a4a935a6a42004ab : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bef36fa1d5ba37a59ee70602a967237385284f59","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-04-18 07:19:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1684896deab04ec8aa48d08f0ca76bf1c27b824f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 07:19:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78355c659eeb24a1c05afc980a2097ff09156e1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 07:48:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0a8887a87a94e818f16d95d2e99763a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff87c4b9b6340f382ab835303bca940 : SUCCESS in 5m 01s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/243352fc2c0c4141ad7493f4bb018d73 : SUCCESS in 10m 15s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/02bf73a1754d414a8e0120d1524023f0 : SUCCESS in 7m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ecc1704731e64e9894ab950f265e80b4 : SUCCESS in 5m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2c7129af153344f89f2ef7ee739a8c21 : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3d1b10536d83f49d1c6e4c63a67dc761c90f188b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-18 07:49:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b34da7c838256f6afc4878d3ebfff65a12d952c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 08:36:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b93255d370a4419a6e16b145ddd65cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c00282708f3e41a6ad4ba05db0b632aa : SUCCESS in 4m 37s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/397696907075467ebebb5454473d932d : SUCCESS in 6m 53s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/11c5e7f55e144846afeb15dbb30e6c9b : SUCCESS in 8m 08s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/bd4a445acc6b41819cf8ffb6dca69b17 : SUCCESS in 8m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4a17dd3914c14527a2bfe8f64f1bc562 : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6389fb8059c121a3178a836a18b173049568d19f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 08:37:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9e8ec6a5830e51498e4a730d56a4ab1592f72b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 09:37:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40de5d334ab44489ab31c6f47f8125be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e5b09afdde4aaf95d885ac675c1733 : SUCCESS in 4m 54s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/149f3e3fa01847b68c271651a1c20c4a : SUCCESS in 8m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c20191a7d7944fdf825fca0a36450ea9 : SUCCESS in 6m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/33945e9b8cd54169aa021408549a8840 : SUCCESS in 5m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1124c76424114964b6bf1fdba1777520 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":1},{"id":"12ed27d70dfc43c7a645cfac94307ba4415f23f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 09:37:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"103cc20c690c5aa69461822c83b0b2daaad1582e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 09:38:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18a064066c0a42a4ba31819c2b9ca05c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80b15f25a214b7c9fd09e6681f8b6d9 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a03ff8ec963e7d5c211864060b826db7b7da595","revisions":{"0a03ff8ec963e7d5c211864060b826db7b7da595":{"kind":"REWORK","_number":1,"created":"2024-04-18 02:19:49.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/53/916153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/53/916153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/53/916153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/53/916153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/53/916153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/53/916153/1"}}},"commit":{"parents":[{"commit":"35f57c4c250b091a5ebdef91213f940a9cf4a1e7","subject":"Merge \"feat: update creating heat stack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/35f57c4c250b091a5ebdef91213f940a9cf4a1e7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-18 02:19:26.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-18 02:19:26.000000000","tz":480},"subject":"fix: update ipv6-encap i18n","message":"fix: update ipv6-encap i18n\n\nupdate ipv6-encap i18n for the security group rule\n\nChange-Id: I067120da290937281fcd1daf2726d6c398251b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a03ff8ec963e7d5c211864060b826db7b7da595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a03ff8ec963e7d5c211864060b826db7b7da595"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
