)]}'
{"id":"openstack%2Fskyline-console~917049","triplet_id":"openstack%2Fskyline-console~master~I46a0ed6e654517593de713bbd0448db505a2138f","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2024-05-08 12:33:39.000000000","reason":"Change was abandoned"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2024-05-08 12:33:39.000000000","reason":"Change was abandoned"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-05-08 12:33:39.000000000","reason":"Change was abandoned"},"7556":{"account":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"last_update":"2024-05-08 12:33:39.000000000","reason":"Change was abandoned"},"36902":{"account":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"last_update":"2024-04-25 14:05:06.000000000","reason":"\u003cGERRIT_ACCOUNT_36902\u003e replied on the change","reason_account":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}},"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2024-05-08 12:33:39.000000000","reason":"Change was abandoned"},"36954":{"account":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"last_update":"2024-05-08 12:33:39.000000000","reason":"Change was abandoned"},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2024-05-07 10:43:20.000000000","reason":"\u003cGERRIT_ACCOUNT_34671\u003e replied on the change","reason_account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}}},"hashtags":[],"change_id":"I46a0ed6e654517593de713bbd0448db505a2138f","subject":"feat: L3 Agents has been added to the detail screen of Routes in the Network tab in admin menu.","status":"ABANDONED","created":"2024-04-25 13:56:25.000000000","updated":"2024-05-08 12:33:39.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"743ee9a055d1f82b0b79ccbb9f32b07c0dabde27","_number":917049,"virtual_id_number":917049,"owner":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-26 14:50:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 14:04:31.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"state":"REVIEWER"},{"updated":"2024-04-25 14:04:31.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"},{"updated":"2024-04-25 14:04:31.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"state":"REVIEWER"},{"updated":"2024-04-25 14:05:06.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2024-04-25 14:05:06.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2024-04-25 14:05:32.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2024-04-25 14:05:45.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"state":"REVIEWER"},{"updated":"2024-04-25 16:33:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f444ffc001e438ca2d05fe22f790db94bfef46d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"date":"2024-04-25 13:56:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01f82abad488d432f6764e19bb3298324611bb0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"date":"2024-04-25 14:08:39.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f34af943a3e6bf72895415d96bf02e454fe0f2bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:33:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48abf1348fbb4f5bb7d29574cdae28f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49528205f8bc4bffad56783f30996a6a : SUCCESS in 4m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0e396d4e66fd4ca5894ed745342d7ae1 : FAILURE in 11m 37s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3a0b21963792450cac9a6576d7ec6277 : SUCCESS in 6m 59s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/12efe7dcc5404eafb9dbe3d322e5b0f7 : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/60a7a565ed354a98924fa8cf09ae264f : FAILURE in 2h 14m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/390873ef56024456b539d964dcae8e2b : FAILURE in 1h 35m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/09705de3d7ec4265b781bb93534dca9b : FAILURE in 1h 25m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/54843dca8a2b4f7d87af980c3b787301 : FAILURE in 1h 31m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5fe5d2629c454f5c304284f76e90a3f89f8eabf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"date":"2024-04-26 12:17:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0f31eac53604dd3b334ccb24b99f1f99bd5793bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 14:50:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0591f8d9fa04ceeb18fc9bab2bb5ca2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f5b2e1b6f3c4d94a04eee61273ca71a : SUCCESS in 4m 14s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/785ab6a28d6040b2a2dd412eb5997e3d : SUCCESS in 7m 21s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/956cfdd5964f4d948fdb6841d569cff5 : SUCCESS in 7m 26s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/fe3de81a8b224060b55b3ff24dbd9c3f : SUCCESS in 7m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4ee4a7a1da52442e91571d2d9ca52656 : FAILURE in 2h 29m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/098b6f80717c48f191e807a97097428e : FAILURE in 1h 28m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cd3f2ebe49f04651a733933b7e2e21c9 : FAILURE in 1h 14m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6b2b58c9311e45ea9ccd749153f40dc1 : FAILURE in 53m 45s","accounts_in_message":[],"_revision_number":3},{"id":"aff2faef88954c420e3c116303021b5d8f5999ec","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2024-04-29 05:59:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a6c32589b32b0264d721e33b9eea7b91fb14a422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 08:13:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd5485ac6acd4f83ae501921c7853242\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd42916132a24544af5bd1e2f47db687 : SUCCESS in 4m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/77cad44b44f24be3a3cae60fb74d11d5 : SUCCESS in 6m 31s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/634ba1aaeea44e11a36f6a491a92e7f0 : SUCCESS in 4m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/87123087639b421e959aea2266c3b3da : SUCCESS in 4m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e322a91815e2453090976a87e92230ad : FAILURE in 2h 13m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b60823be9265476a9fe55e5f789e79a4 : FAILURE in 1h 26m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e0a7b955b48a451199e2b28524818f2f : FAILURE in 1h 13m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/30fa73f75cdb4ed394366d35771a8202 : FAILURE in 51m 11s","accounts_in_message":[],"_revision_number":3},{"id":"00c7076c2042d9345e26f5f78fb2f00c42032eaa","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-05-07 10:43:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b92f5b432ade9eb9df5369a50508b9b779050d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 13:53:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93fc45f8e2be46f9ac06e1c4cd5f5264\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36113370aea48309ff4d61236773909 : SUCCESS in 4m 24s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/450755342c544755b309115a537ec04d : SUCCESS in 10m 54s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/46994979bb7b419a8ddaebcbc164022b : SUCCESS in 7m 45s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/faf6d4da5e0342cba543373281db24e4 : SUCCESS in 6m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b73a3212f18f4d8c8562e137c5bd0ca9 : TIMED_OUT in 3h 03m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c6b093e77c99407a9dbc095b586622aa : FAILURE in 1h 42m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/38f42078e69549d590060075e67bfe90 : FAILURE in 1h 14m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/90ec969e45d741e683385732010f4dd2 : FAILURE in 59m 45s","accounts_in_message":[],"_revision_number":3},{"id":"743ee9a055d1f82b0b79ccbb9f32b07c0dabde27","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"date":"2024-05-08 12:33:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"308bdcb7eb4e5cd74d3491397b76a362477e4d7b","revisions":{"786c64d30af7e21d4bb400ec068d4a678a566a99":{"kind":"REWORK","_number":1,"created":"2024-04-25 13:56:25.000000000","uploader":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"ref":"refs/changes/49/917049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/49/917049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/1"}}},"commit":{"parents":[{"commit":"23a74052a56597abc5349206bec2366455d22c1e","subject":"Merge \"fix: update ipv6-encap i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23a74052a56597abc5349206bec2366455d22c1e"}]}],"author":{"name":"nurmermer","email":"nur.mermer@tubitak.gov.tr","date":"2024-04-25 07:11:43.000000000","tz":0},"committer":{"name":"nurmermer","email":"nur.mermer@tubitak.gov.tr","date":"2024-04-25 13:56:27.000000000","tz":0},"subject":"feat: L3 Agents has been added to the detail screen of Routes in the Network tab in admin menu.","message":"feat: L3 Agents has been added to the detail screen of Routes in the Network tab in admin menu.\n\nChange-Id: I46a0ed6e654517593de713bbd0448db505a2138f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/786c64d30af7e21d4bb400ec068d4a678a566a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/786c64d30af7e21d4bb400ec068d4a678a566a99"}]},"branch":"refs/heads/master"},"1d9b4b6dca004c684604b7c8d65761e3b80df5a8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-25 14:08:39.000000000","uploader":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"ref":"refs/changes/49/917049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/49/917049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/2"}}},"commit":{"parents":[{"commit":"23a74052a56597abc5349206bec2366455d22c1e","subject":"Merge \"fix: update ipv6-encap i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23a74052a56597abc5349206bec2366455d22c1e"}]}],"author":{"name":"nurmermer","email":"nur.mermer@tubitak.gov.tr","date":"2024-04-25 07:11:43.000000000","tz":0},"committer":{"name":"İrem Nur Mermer","email":"nur.mermer@tubitak.gov.tr","date":"2024-04-25 14:08:39.000000000","tz":0},"subject":"feat: L3 Agents has been added to the detail screen of Routes in the Network tab in admin menu.","message":"feat: L3 Agents has been added to the detail screen of Routes in the Network tab in admin menu.\n\nL3 Agents which belong to each routes detail has been added to the Routes detail screen in admin menu.\n\nChange-Id: I46a0ed6e654517593de713bbd0448db505a2138f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1d9b4b6dca004c684604b7c8d65761e3b80df5a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1d9b4b6dca004c684604b7c8d65761e3b80df5a8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"308bdcb7eb4e5cd74d3491397b76a362477e4d7b":{"kind":"REWORK","_number":3,"created":"2024-04-26 12:17:34.000000000","uploader":{"_account_id":36954,"name":"İrem Nur Mermer","display_name":"iremnur","email":"nur.mermer@tubitak.gov.tr","username":"nurmermer"},"ref":"refs/changes/49/917049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/49/917049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/49/917049/3"}}},"commit":{"parents":[{"commit":"23a74052a56597abc5349206bec2366455d22c1e","subject":"Merge \"fix: update ipv6-encap i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23a74052a56597abc5349206bec2366455d22c1e"}]}],"author":{"name":"nurmermer","email":"nur.mermer@tubitak.gov.tr","date":"2024-04-25 07:11:43.000000000","tz":0},"committer":{"name":"nurmermer","email":"nur.mermer@tubitak.gov.tr","date":"2024-04-26 12:15:16.000000000","tz":0},"subject":"feat: L3 Agents has been added to the detail screen of Routes in the Network tab in admin menu.","message":"feat: L3 Agents has been added to the detail screen of Routes in the Network tab in admin menu.\n\nChange-Id: I46a0ed6e654517593de713bbd0448db505a2138f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/308bdcb7eb4e5cd74d3491397b76a362477e4d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/308bdcb7eb4e5cd74d3491397b76a362477e4d7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
