)]}'
{"id":"openstack%2Fskyline-console~918379","triplet_id":"openstack%2Fskyline-console~master~I24d1e831429bba2ce118d42003dae723254045b2","project":"openstack/skyline-console","branch":"master","topic":"application-credential","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-05-07 06:19:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24d1e831429bba2ce118d42003dae723254045b2","subject":"feat: update prompts for unrestricted","status":"MERGED","created":"2024-05-07 04:25:31.000000000","updated":"2024-05-07 06:20:26.000000000","submitted":"2024-05-07 06:19:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918379-application-credential","meta_rev_id":"f2f1238ccfe32ff6b11b34f6a6383d7de22e305b","_number":918379,"virtual_id_number":918379,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-07 06:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-05-07 05:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-05-07 05:35:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-07 05:06:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 05:35:49.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"911ad18c074f101f9e18917396d85d7bb44d5a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-05-07 04:25:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f541121be3851a90d37c837c7a7b8ba57a61ef5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 05:06:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76e162b0614a4d9f83396773edb33933\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c091737ae51412e8b4ed251de9ecd3c : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a8bc0feb514587abd8bd1581afcdef : SUCCESS in 4m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4b49ff331d664793b96873aff6fbd8a2 : SUCCESS in 8m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/296f31c95cde4574937966e6ba9609af : SUCCESS in 6m 54s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/85fbfcdb96ce479bbbbe18c62605985f : SUCCESS in 6m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/78061310a02b45d5b8bf3a2ad908b49b : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":1},{"id":"352ad9613122a4f074d7c0c805f854e3196947a3","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-05-07 05:35:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2037edcc23ba9186224f56ba8687df2eb637bfb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 05:36:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cae0177e9068b2f1d9dd1e387bacb4befbd3326f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 06:19:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85366b8997f944efaab55bcc93cedcb3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/732262d6796845358cf2b48042fcbcaf : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3737e763ca425b8818b76d4927ca83 : SUCCESS in 5m 14s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/79a0e299708a4f6a9809865fa52416af : SUCCESS in 6m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c6668486ab2647f58572bb267c93be77 : SUCCESS in 6m 51s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1182b1aaae7741fcaa2cf9cb5bd6a623 : SUCCESS in 4m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/055f24cdf9a2437eb49822f3256e90bb : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":1},{"id":"decfba796f68036ff747c44c9640d5653b1f86c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 06:19:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2f1238ccfe32ff6b11b34f6a6383d7de22e305b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 06:20:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5aad82f513c423bb6287a56c88a22eb\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dd1b07fe3874cda8320b613ba64b17a : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d802ffec51f422480673d3b56ca12f9 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6ebeee2d43fcdd6f37cd09c564d15cbff59da32","revisions":{"e6ebeee2d43fcdd6f37cd09c564d15cbff59da32":{"kind":"REWORK","_number":1,"created":"2024-05-07 04:25:31.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/79/918379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/918379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/918379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/918379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/918379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/918379/1"}}},"commit":{"parents":[{"commit":"027898907f6ced87c255f7700e0d8a8cb2583ead","subject":"Merge \"feat: update refresh tab data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/027898907f6ced87c255f7700e0d8a8cb2583ead"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-05-07 04:25:03.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-05-07 04:25:03.000000000","tz":480},"subject":"feat: update prompts for unrestricted","message":"feat: update prompts for unrestricted\n\nUpdate the prompts for the unrestricted attribute when creating application credential.\n\nChange-Id: I24d1e831429bba2ce118d42003dae723254045b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e6ebeee2d43fcdd6f37cd09c564d15cbff59da32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e6ebeee2d43fcdd6f37cd09c564d15cbff59da32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
