)]}'
{"id":"openstack%2Fskyline-console~918707","triplet_id":"openstack%2Fskyline-console~master~I983b9ca2729ba8e762d67d42186a481b7224e18b","project":"openstack/skyline-console","branch":"master","topic":"firewall","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-05-09 07:02:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I983b9ca2729ba8e762d67d42186a481b7224e18b","subject":"test: add firewall e2e","status":"MERGED","created":"2024-05-09 02:58:29.000000000","updated":"2024-05-09 07:03:25.000000000","submitted":"2024-05-09 07:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918707-firewall","meta_rev_id":"23967f697144175036cfa886fcebb9421857205a","_number":918707,"virtual_id_number":918707,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 07:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-05-09 05:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-05-09 05:32:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-09 04:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-09 05:32:29.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"4f8e9875651ba7a9f74161413958ea7c30208288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-05-09 02:58:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1250f9e5d4baf5e36cc997c615980b08f3799131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 04:06:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffcc5cc1193d4397bd5dcc32c917ed54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14affaca564a4e9bb9490d3e8376e4eb : SUCCESS in 4m 41s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d9dd791c55ed4a24b5d4b6bc1e40b39a : SUCCESS in 9m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/404bfee7461141d684544c8382d47ffb : SUCCESS in 11m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3a069b4d433f40c38bf6a8a994fc8518 : SUCCESS in 4m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bc44c92d033a48b59ad3683e22b1b71f : SUCCESS in 1h 06m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/72cc4fb6f3b949b096f8f678ccc0ecce : SUCCESS in 50m 57s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e8c44e193a6e4f7bbc33d4e35ffbcecb : SUCCESS in 46m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2beaae6eb4414c75b5c626d4bf7decf2 : SUCCESS in 43m 38s","accounts_in_message":[],"_revision_number":1},{"id":"835cde46ec37f4c60e86b759bbafa19200162d33","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-05-09 05:32:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4c685c2b82d0b2f976ed29d95c00a6c12d4ba5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 05:33:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33fe38e9da77e3496b9a35c342136cccfedb2c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:02:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab838373dc664ea3bcf64958435463b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86af2b5092ef4024a02c45d31bbc257c : SUCCESS in 4m 09s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f46ddd2b51d24bc39d430dd44206d325 : SUCCESS in 10m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2d8fa3b82e3c4ac0baecd81122665c80 : SUCCESS in 4m 40s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0aa66de7778c46ffbe26d86bbfa4d706 : SUCCESS in 7m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3aaae6e208544fa18d0e9ed7c773877b : SUCCESS in 1h 23m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1b622267191a42b1b0deab39c32d4b44 : SUCCESS in 44m 19s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f3b4c08a01524d04853a9f25277e884b : SUCCESS in 40m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a518d922326f42b793a8b970e44bfe8a : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":1},{"id":"27e265c145cd88afb997cdb5d9bda3fc403b453f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:02:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23967f697144175036cfa886fcebb9421857205a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:03:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/238d535d28d04713b60304a6d14eb3aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47696b938512435cb66164f840e1ea20 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d8d8a76d94f249db176f576e110a3e69bf77307","revisions":{"2d8d8a76d94f249db176f576e110a3e69bf77307":{"kind":"REWORK","_number":1,"created":"2024-05-09 02:58:29.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/07/918707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/918707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/918707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/918707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/918707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/918707/1"}}},"commit":{"parents":[{"commit":"5c129f2705ac7790a8b8b4e80f1037ddb3860468","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5c129f2705ac7790a8b8b4e80f1037ddb3860468"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-05-09 01:26:17.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2024-05-09 02:57:27.000000000","tz":480},"subject":"test: add firewall e2e","message":"test: add firewall e2e\n\nadd e2e cases for firwall/firewall-policy/firewall-rule\n\nChange-Id: I983b9ca2729ba8e762d67d42186a481b7224e18b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d8d8a76d94f249db176f576e110a3e69bf77307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d8d8a76d94f249db176f576e110a3e69bf77307"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
