)]}'
{"id":"openstack%2Fskyline-console~918709","triplet_id":"openstack%2Fskyline-console~master~Iee60a49733514f52c5a1030eeec382d3e4beaf3b","project":"openstack/skyline-console","branch":"master","topic":"monitor-openstack","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-05-09 07:28:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee60a49733514f52c5a1030eeec382d3e4beaf3b","subject":"feat: add instance info","status":"MERGED","created":"2024-05-09 04:11:25.000000000","updated":"2024-05-09 07:29:27.000000000","submitted":"2024-05-09 07:28:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918709-monitor-openstack","meta_rev_id":"bd04d8184b8f026569b386b50184de805e12aff6","_number":918709,"virtual_id_number":918709,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 07:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-05-09 05:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-05-09 05:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-09 05:35:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-09 05:50:00.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"60500c10877f6e4eb1f6615b1326f6981202a5b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-05-09 04:11:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8bf0c806c42656c611cba71d78a2c909d15c722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 05:35:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e5fb2e55e84e42978deaff9173be45\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01024a146c9a4118ababb2d2a0f5199f : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc2f7350ffa4a91a3085e9d5042a881 : SUCCESS in 4m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7bff44c2330c43dcb3dec22092f4e06a : SUCCESS in 10m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4e96c521d98f4f52b22d6212619af4e9 : SUCCESS in 7m 07s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4a620e1eff824c2fa20cb0f0b4cf8778 : SUCCESS in 7m 25s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d51d1ae31917490ba4fd4a0694d49fae : SUCCESS in 1h 18m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f926b56da15145369e66e4d369b9408a : SUCCESS in 59m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3d0ed0ef20de4834af770664e323fab7 : SUCCESS in 53m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7c4f5f7dbcfb4202b931e9aa498e1eba : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"aae91818ae877edec6646dbb5bcb571178bc0c0c","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-05-09 05:50:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"062bdbe64e35cccdc1f8662e839fa27077bab556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 05:50:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7945e1db10e1f6160b99807d6168b89b4e582f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:28:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7234c70fe0c477d818678d3fd6ac6e2\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dba4edb6710449aba5ddb30972b0817f : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869693895ba7459a87fb9ef4b6f1782a : SUCCESS in 4m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d3ed90ebf83c4a7891c5f3a700063e25 : SUCCESS in 10m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/287abff54b2d470d8666dd155395cbd0 : SUCCESS in 7m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/93e2e9bdede1439187de063a51769567 : SUCCESS in 6m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/81555258eb2b4a19b2b61316a964bb96 : SUCCESS in 1h 32m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4db9fbcd9f054e3aa0c17c0ca713c704 : SUCCESS in 1h 03m 55s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/adb25d64fe714464b89c4ba6887c53bd : SUCCESS in 1h 00m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a3556cf1f01348a398d78bf15d1becce : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"75885d914fba94a23f28221bf8d587c5ada94df7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:28:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bd04d8184b8f026569b386b50184de805e12aff6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:29:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b7465fce6ce4e8c987584367f7ea346\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae0e673c2d2547fdad8a40ba80aa01f0 : SUCCESS in 49s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d315d8579bbe4ac2899739f90f2b8e4b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f4c8fdb9572bf827856ad048d246ffcaeaf4621","revisions":{"3f4c8fdb9572bf827856ad048d246ffcaeaf4621":{"kind":"REWORK","_number":1,"created":"2024-05-09 04:11:25.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/09/918709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/918709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/918709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/918709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/918709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/918709/1"}}},"commit":{"parents":[{"commit":"5c129f2705ac7790a8b8b4e80f1037ddb3860468","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5c129f2705ac7790a8b8b4e80f1037ddb3860468"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-05-09 04:09:49.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-05-09 04:11:05.000000000","tz":480},"subject":"feat: add instance info","message":"feat: add instance info\n\nAdd instance info for the openstack service monitor in the administrator\n\nChange-Id: Iee60a49733514f52c5a1030eeec382d3e4beaf3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f4c8fdb9572bf827856ad048d246ffcaeaf4621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f4c8fdb9572bf827856ad048d246ffcaeaf4621"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
