)]}'
{"id":"openstack%2Fskyline-console~923347","triplet_id":"openstack%2Fskyline-console~master~Ief764ce04f61eea06fd6274b55b41fe19f9f32ca","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-07-03 07:44:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief764ce04f61eea06fd6274b55b41fe19f9f32ca","subject":"fix: update row action button style","status":"MERGED","created":"2024-07-03 02:30:05.000000000","updated":"2024-07-03 07:45:30.000000000","submitted":"2024-07-03 07:44:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923347","meta_rev_id":"00da5893b814f565e51db25c44540c1f48b22db4","_number":923347,"virtual_id_number":923347,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-03 07:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-07-03 06:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-07-03 06:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 04:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 06:26:54.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"da0067521e7a254515aed9ef2ba95182a55a1fe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-07-03 02:30:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7130f5464a4146cb0213452f513f5b1227a93f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 04:48:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebc88ae7158a4d5385c5d5b4d877fedf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06f7a5f0de3842cfa9a1dc76e088b36b : SUCCESS in 5m 44s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/75907e23e2134dc698fcfeea4f84cd90 : SUCCESS in 8m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a619a71b4edb4378a1edfe4353850d0f : SUCCESS in 7m 21s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1ec2e67a35d9406bae553e1b912b2be0 : SUCCESS in 7m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1305423332fb4110a0c757a0b3212df5 : SUCCESS in 1h 14m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8622a75213ef475a9fe72b5ee025642a : SUCCESS in 51m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/884f1734eed14c09a71f323cd9abd9f1 : SUCCESS in 43m 41s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f078094d039941bf895b057ff37e60fa : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e55e11e98201ae72b260b28064ae9ebecd4a8586","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-07-03 06:26:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94cebb5e033b5acaed4fc3628266083571c9eb31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 06:27:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f88999b426c93de8b3558219c4f6c76541827d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 07:44:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcf739d2f9674888ac5293ffb38d263c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d09ba46b4454444c80a40005551804e4 : SUCCESS in 4m 40s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4b866f1aaee744c1bc9c595e433b5bff : SUCCESS in 8m 27s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0580a9288536469f9e4c42668ddeac5b : SUCCESS in 7m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e589b8a703c64af8b8068deab81e9524 : SUCCESS in 7m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3c3f5a38fad843738b11f17d7b6dbf3b : SUCCESS in 1h 14m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1350b94b988c445895af4e86cc2766fc : SUCCESS in 51m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/89bcbdc4e0ed49df8d74348725aea74d : SUCCESS in 59m 10s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6771721f0b9e4a0bb7f26ac4a1bcd161 : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8249f3dc5e3ba12678e6ae63d40a7ea88377e71b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 07:44:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"00da5893b814f565e51db25c44540c1f48b22db4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 07:45:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa010145fba947d3b1e9279256e5783b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4d3ed4b13a40cfacdf01aa25e08334 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26ed30cd81705acf533636edb0b93bfbee67b69d","revisions":{"26ed30cd81705acf533636edb0b93bfbee67b69d":{"kind":"REWORK","_number":1,"created":"2024-07-03 02:30:05.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/47/923347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/923347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/923347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/923347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/923347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/923347/1"}}},"commit":{"parents":[{"commit":"0b7d79f9dd4107258e7e5940ff2ae6f88f3eb1b3","subject":"Implements: fix bug, check type of prometheus query result data Closes-Bug: 2015991","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0b7d79f9dd4107258e7e5940ff2ae6f88f3eb1b3"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-07-03 02:29:18.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-07-03 02:29:27.000000000","tz":480},"subject":"fix: update row action button style","message":"fix: update row action button style\n\n1. update the style of the row action buttons in the list page\n2. update the style of the action buttons at the top of the detail page\n\nChange-Id: Ief764ce04f61eea06fd6274b55b41fe19f9f32ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/26ed30cd81705acf533636edb0b93bfbee67b69d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/26ed30cd81705acf533636edb0b93bfbee67b69d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
