)]}'
{"id":"openstack%2Fskyline-console~924151","triplet_id":"openstack%2Fskyline-console~stable%2F2024.1~Id93fe64e83c2951b069d97177f6da851271a7ba8","project":"openstack/skyline-console","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2024-09-10 01:30:20.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}},"hashtags":[],"change_id":"Id93fe64e83c2951b069d97177f6da851271a7ba8","subject":"Lock setuptools version as 69.2.0","status":"MERGED","created":"2024-07-15 13:20:44.000000000","updated":"2024-09-13 06:38:19.000000000","submitted":"2024-09-13 06:37:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924151","cherry_pick_of_change":924130,"cherry_pick_of_patch_set":1,"meta_rev_id":"7dfe3616bbcefc2cce64aea1b555625ae093c203","_number":924151,"virtual_id_number":924151,"owner":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-13 06:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-09-13 05:41:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-09-13 05:41:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-15 16:07:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91ad087e5120589c2a93000530b131a7c87f551f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-07-15 13:20:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"205c2e21b9c42c56f7e65395867caf82c985751d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 16:07:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c42344a24ecc4e18bd617816145d61bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d21c94e74b4e5a9c2076ff803a8616 : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/96241b1da0df429ca78bad804cdaa544 : SUCCESS in 9m 40s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/27e5c9c69adb40a4862423804a760f43 : SUCCESS in 6m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3f25b4130398459d80c47bb6c312c9f3 : SUCCESS in 5m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1b8fb6f4034d44be9d7376c6a81fd985 : SUCCESS in 1h 02m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/183eade8ed29473983795725f67e905c : SUCCESS in 1h 04m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0ac8166cf39b4cedb21c777035458fd4 : FAILURE in 59m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/eb33cb78a8164a378d2a11b839f36c19 : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f724820c631247e7f4f64d2a1beed1dc37486f2b","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-09-09 08:28:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc3ede7112bd5c734c44a5b46df91f95441a8b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 09:32:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/633b88012a5244e9be47ae6b00d0585f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2061b652f2f4005b5fd3ea4253df529 : SUCCESS in 4m 29s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3b5ccb992d4741f4b7c977f39ffac075 : RETRY_LIMIT in 2m 20s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d92c0d16f7af4bde9348728928f6a46b : RETRY_LIMIT in 2m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/eb11c2d603474f16a61c31923538bb2b : RETRY_LIMIT in 3m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/938c2ce34ed546acabd454a665078a33 : SUCCESS in 55m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fcc259e822e441eebc31387d0874b272 : SUCCESS in 58m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/22f889c8f45b4fc0a5a5ae353259681e : SUCCESS in 51m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/db6c6178408c4ff4bbaaff0ded340af8 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cadd2244c0421903b416c159577c63fe63eb9d2f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-09-10 01:30:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c594bad58a07668ea6fa41fab7da7c8eddd9f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 02:36:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6890968898c4cc2b7e6ce3b1eaafa5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563a1e5dee9041c5be22a68d319f51ff : SUCCESS in 4m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ffa1571a73104cc4a8feb03b253a6e59 : SUCCESS in 9m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f9e6aec652304368bac25aa47a619c57 : SUCCESS in 6m 53s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/61dd8a75439b49058a584db5fad6038e : SUCCESS in 4m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ee105957c9b144d3a851f857d3ff9cf4 : SUCCESS in 1h 03m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9b714483fb644ff9b3e37d666f9ed8de : SUCCESS in 51m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/03f26e1c5fba4509ae6d42997fc64ba5 : SUCCESS in 1h 00m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b2a2d389a619474db9a32d563c089512 : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a361736c372f6ba65950ce84fd683d9bf6181593","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-09-13 05:41:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0529e55e99019a18d24a30fc06dbde7558c05e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 05:42:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2e8690a92397f1aa197393c167666fddf9b0ba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 06:37:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a854d6fbf5cf4d819c75e9625c3f5a7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e126bb56c0bc4fd3bee40af0ec4f1392 : SUCCESS in 4m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/62147b52412542018763e184ab592098 : SUCCESS in 11m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6beb4258549c4886a268ab5535c4ef2a : SUCCESS in 3m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/96e237199b004db5b6616be9d3023966 : SUCCESS in 6m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5de7e7c1583b4056bbca973295ac32ed : SUCCESS in 53m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b92774383552464ca76cada2a0591f82 : SUCCESS in 51m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d92b61bf2453465393ce9a833a22156b : SUCCESS in 45m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4e4bc29d0af844f58b0926228a0e795b : SUCCESS in 44m 10s","accounts_in_message":[],"_revision_number":1},{"id":"85728e4c508709d4b846084dc5f1cdb948a4aefa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 06:37:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7dfe3616bbcefc2cce64aea1b555625ae093c203","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 06:38:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8abf899c787413b99b20b6c4eefcc58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94bffa4a3b64f3d99628f2210a14ee7 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f4d1e01b1ec48334ea79de0660177004c891b52","revisions":{"7f4d1e01b1ec48334ea79de0660177004c891b52":{"kind":"REWORK","_number":1,"created":"2024-07-15 13:20:44.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/51/924151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/924151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/924151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/924151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/924151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/924151/1"}}},"commit":{"parents":[{"commit":"667b8559f95184a5cdf55ccbe1f4806bc31c2df8","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2024.1\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/667b8559f95184a5cdf55ccbe1f4806bc31c2df8"}]}],"author":{"name":"zhuboxiang","email":"zhu.boxiang@99cloud.net","date":"2024-07-15 08:34:10.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2024-07-15 13:20:44.000000000","tz":0},"subject":"Lock setuptools version as 69.2.0","message":"Lock setuptools version as 69.2.0\n\nLock setuptools version as 69.2.0. We need package like\nskyline-console tar.gz. If we use higher version, we will\nget skyline_console tar.gz package.\n\nChange-Id: Id93fe64e83c2951b069d97177f6da851271a7ba8\n(cherry picked from commit f70772ecc89f076525bd377fe711fcd1d4d97678)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7f4d1e01b1ec48334ea79de0660177004c891b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7f4d1e01b1ec48334ea79de0660177004c891b52"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
