)]}'
{"id":"openstack%2Fskyline-console~924897","triplet_id":"openstack%2Fskyline-console~master~I300a3020b8402ca9638e1139a14c460a6a0a2660","project":"openstack/skyline-console","branch":"master","topic":"instance","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-07-25 07:06:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I300a3020b8402ca9638e1139a14c460a6a0a2660","subject":"fix: fix edit instance","status":"MERGED","created":"2024-07-25 02:53:36.000000000","updated":"2024-07-25 07:07:35.000000000","submitted":"2024-07-25 07:06:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924897-instance","meta_rev_id":"82b3c2bf2940ebc970bb0eac1d76150a91e8a3b0","_number":924897,"virtual_id_number":924897,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-25 07:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-07-25 05:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-07-25 05:40:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-25 04:45:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-25 05:40:16.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"439b963e4ddf84bcd34fb579917d66e7b0b4bf5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-07-25 02:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcd70703fc49ef6b36dcd610bb80636c47b48375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 04:45:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caa878d49c87408bba8dbc11ef181fb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4a893632074246beab177ee3abff09 : SUCCESS in 4m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f88fa24a7ff04cebb477fde6aee37d2e : SUCCESS in 10m 37s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/5eb377214f9a499bb9bfebde24e82b7b : SUCCESS in 7m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/97c138d93aa642698494714c60fe04a5 : SUCCESS in 6m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/14b792886a3b4771b0b66cfa4c5cfdc5 : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ec9290bf50f85b789f42798ffc2017c6be2465b2","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-07-25 05:40:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcc56dba16ff43f3d99701f14c819b8e153a3064","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 05:40:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5866392943ab2daa7da6a01b28c87bd59216ffa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 07:06:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5f78a0753444851a50f4494ab7a95d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc2e330e51a54759899687553104d7ff : SUCCESS in 5m 29s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/adc7a4a6638b4385aa11aeac28c2a0d0 : SUCCESS in 8m 29s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0e6a057500234334b2617fca05435ead : SUCCESS in 7m 16s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/31ece614d7cd4200845f5906b5d7f8cf : SUCCESS in 6m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8c7d1d4a19d94490a07ccdaf115bf451 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a49c09409832d5c41d24c109a00d1e7f261033bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 07:06:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"82b3c2bf2940ebc970bb0eac1d76150a91e8a3b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 07:07:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa15ca90d706473494ecc676bb970724\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/165a0ca1d8924270a927b7274c6aca5c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7513e64f7c1d7a568539f89f1953c859d938aa3d","revisions":{"7513e64f7c1d7a568539f89f1953c859d938aa3d":{"kind":"REWORK","_number":1,"created":"2024-07-25 02:53:36.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/97/924897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/97/924897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/924897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/924897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/924897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/97/924897/1"}}},"commit":{"parents":[{"commit":"f70772ecc89f076525bd377fe711fcd1d4d97678","subject":"Lock setuptools version as 69.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f70772ecc89f076525bd377fe711fcd1d4d97678"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-07-25 02:51:49.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-07-25 02:52:58.000000000","tz":480},"subject":"fix: fix edit instance","message":"fix: fix edit instance\n\nfix edit instance in the instance detail page for the non-admin user\n\nChange-Id: I300a3020b8402ca9638e1139a14c460a6a0a2660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7513e64f7c1d7a568539f89f1953c859d938aa3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7513e64f7c1d7a568539f89f1953c859d938aa3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
