)]}'
{"id":"openstack%2Fskyline-console~929249","triplet_id":"openstack%2Fskyline-console~stable%2F2024.2~I3e897d24a5d86df22d407e52c4dfce78afd2ba96","project":"openstack/skyline-console","branch":"stable/2024.2","topic":"create-2024.2","hashtags":[],"change_id":"I3e897d24a5d86df22d407e52c4dfce78afd2ba96","subject":"Update .gitreview for stable/2024.2","status":"MERGED","created":"2024-09-13 11:12:47.000000000","updated":"2024-10-26 13:42:49.000000000","submitted":"2024-10-26 13:41:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929249-create-2024.2","meta_rev_id":"b789bad6e1c1184bc84c928a57407b9baaf3c26a","_number":929249,"virtual_id_number":929249,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-26 13:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-10-26 12:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2024-09-27 14:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-10-26 12:17:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-13 11:39:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-27 14:12:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-10-26 12:17:10.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"77973bed2d32720e62d45407f90b9dbf0d4213b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-09-13 11:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02ce7275bddf038031ab7daacb904cbf7f2e83d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 11:39:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74f7d0e9e0654c089734324a150779ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a2aa95a6c84fcba411b935044ee15d : SUCCESS in 4m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/48193bb4d1b5405d8f01eb579129ff0d : SUCCESS in 7m 55s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/525f3c0997ab446a9c36487ee0de9797 : SUCCESS in 4m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6780aafe57e640e98d51c6a0fd3a8dee : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"761841993f99907ae34fe624737153aa9089d27d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-27 14:12:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b92495c3a611e1bd025609440b1dde7eeffd8b46","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-10-26 12:17:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5da4e7c3ba266a0dff866aaf39515c37ebcc4109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 12:18:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6abbfa4f30a4b30acb1b83213f3004270d0b8322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 13:41:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338b5b3c769f41b8b51df98f358ab03b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc62d5a9d0740a38f122f6cedfc2834 : SUCCESS in 3m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c0a74a321782404d9a813562d5e4eb12 : SUCCESS in 4m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/95501500e50947a0b56d22a95a134c49 : SUCCESS in 5m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f96f9a7de59f4f2492f00a9de8d8e248 : SUCCESS in 5m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/05d0810ea32b47e9af691f6d9e7b0131 : SUCCESS in 1h 14m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e641a5ef5b5c4ce89e56aeaaaab5df9e : SUCCESS in 52m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f1029e26c9384352b50aa2475872a95f : SUCCESS in 44m 06s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e2b1a1224fa34923b825c73a99b799ea : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":1},{"id":"52b75f6066c99f1715432d411b7afebca7c75915","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 13:41:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b789bad6e1c1184bc84c928a57407b9baaf3c26a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 13:42:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6e77def6274ede8499f6f316d68e41\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89520068334482a89a9845656691d07 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1cef14497dc043acd9924a4966febe7577c1b5c","revisions":{"a1cef14497dc043acd9924a4966febe7577c1b5c":{"kind":"REWORK","_number":1,"created":"2024-09-13 11:12:47.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/49/929249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/49/929249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/929249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/929249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/929249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/49/929249/1"}}},"commit":{"parents":[{"commit":"d037b061f41758023f2434b4752e866da766fe03","subject":"fix: fix user/group column","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d037b061f41758023f2434b4752e866da766fe03"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-13 11:12:46.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-13 11:12:46.000000000","tz":0},"subject":"Update .gitreview for stable/2024.2","message":"Update .gitreview for stable/2024.2\n\nChange-Id: I3e897d24a5d86df22d407e52c4dfce78afd2ba96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1cef14497dc043acd9924a4966febe7577c1b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1cef14497dc043acd9924a4966febe7577c1b5c"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
