)]}'
{"id":"openstack%2Fskyline-console~930534","triplet_id":"openstack%2Fskyline-console~stable%2F2023.2~If2e93c1fa50f6745e7b670f01d212bd8872c2b5f","project":"openstack/skyline-console","branch":"stable/2023.2","topic":"doc/align-container-tag-with-branch","attention_set":{},"removed_from_attention_set":{"33786":{"account":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"last_update":"2024-12-24 02:30:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If2e93c1fa50f6745e7b670f01d212bd8872c2b5f","subject":"Fix container tag for 2023.2 on doc","status":"MERGED","created":"2024-09-26 06:06:17.000000000","updated":"2024-12-24 02:31:54.000000000","submitted":"2024-12-24 02:30:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930534-doc/align-container-tag-with-branch","meta_rev_id":"b654be6a5c11e6c1edb85ac698fed5217f20d242","_number":930534,"virtual_id_number":930534,"owner":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-24 02:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-12-24 02:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-12-24 02:04:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 06:32:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-24 02:04:12.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"944d2a7628d3d86b2e4665aec6942603e2592dc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 06:06:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"955f01951e2be3d0e864868d0f933a9ba8ee2535","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 06:07:23.000000000","message":"Topic topic/Align-container-tag-with-branch removed","accounts_in_message":[],"_revision_number":1},{"id":"22dd5b9c1bcb250f75f9fdb7a2d6fd3da94b7663","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 06:07:44.000000000","message":"Topic set to doc/Align-container-tag-with-branch","accounts_in_message":[],"_revision_number":1},{"id":"0b8c653ce632627634c22f3e6b18623cf6a2ae66","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 06:29:00.000000000","message":"Topic doc/Align-container-tag-with-branch removed","accounts_in_message":[],"_revision_number":1},{"id":"57293ab7197301b7bdffff39a4b4cfc6cde43910","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 06:29:14.000000000","message":"Topic set to doc/align-container-tag-with-branch","accounts_in_message":[],"_revision_number":1},{"id":"fc0c62237f339a3e16c163a6c5e0c36952483117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 06:32:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ceb50b3f57c14ae8af3335d1e3cdc3bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7738445ea739419b96b041a7d9ce3d19 : SUCCESS in 4m 14s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d51de659dfe947d2b876c9511061bb60 : RETRY_LIMIT in 2m 05s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/16f62d8cfeff4b2a81c0925a2efd2672 : RETRY_LIMIT in 2m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3d8eff201fe9487da791c8dad3a53537 : RETRY_LIMIT in 2m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0ac8ced9ea502af7894f76321dc1a56d20110ceb","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 10:01:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc2d3d95126f55dacddc7da5053da6d1ab726235","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-27 01:01:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88905ea2bc918476ce15691fff52ac39d629122f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 01:26:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/706ab45be2334796b0230186afde1f17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c652f8f85852425787e7f823d7f6d836 : SUCCESS in 4m 06s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4f44a6ee21344f5181a7e46d89ff6613 : RETRY_LIMIT in 2m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0fe417838dc44856968f6c484c1b4f05 : RETRY_LIMIT in 2m 10s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7d96a9ff0e2046ffb92718210ca0c3ab : RETRY_LIMIT in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"76588d9581a44357f7e543362bf79c30d2db0cd6","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-10-08 07:30:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91c7aa891ecc3b4f25ede15d919681031b950426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 08:02:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73aa87cfb3da4e39ae8d421642c54c59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e37d2869f2d1450f879b94759c3a8b7d : SUCCESS in 4m 44s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/10c9be890e7745a8ae3e4ff5b11a61cd : RETRY_LIMIT in 2m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/5b66f159185249daac1d89f01f392342 : RETRY_LIMIT in 2m 15s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/bc5c31281b314fa0a51c1090ad3e5c89 : RETRY_LIMIT in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"755efd68d3903fea035b9dc25c0bd29559b7319e","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-12-20 05:12:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6a7377d9f18c16a749d0011f7ba17f37900b409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 05:30:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db5f69bb37c44637abe33ae6a4f0f8ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c413114609a841f885539ce597fff453 : SUCCESS in 3m 49s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7317520d50314b179f9a6dd4d60defa4 : RETRY_LIMIT in 2m 02s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2c3a607e204e4e1abc9cb3f1bdcd1615 : RETRY_LIMIT in 2m 06s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/fba4746665d04a9ba025ac1195aae904 : RETRY_LIMIT in 2m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ef7ae2f0025c2c1c981c4385eda8a6f8ac27ecdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"real_author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-12-23 09:48:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_33786\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"}],"_revision_number":2},{"id":"945e71c0b46f0b4afef3ba052c199742153d2047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 09:57:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78a3cb6e15c4e67a69aea956e7f163c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce7c0499114421985194ba5e89c2a9f : SUCCESS in 4m 31s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/aeb35a88380c4d9eb85c5e8a24cbfc7f : SUCCESS in 7m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/115385ae46e7429ea7a167ae6ff7f8e2 : SUCCESS in 4m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/457340c4c9484ec99dff9bf1247fa3f5 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"02cd6750b64e8df5d70b133409c17fac57f86b60","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-12-24 02:04:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"22f4095a53460e77e4bb2690aa077239f04fa4a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 02:15:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff382db3b444225f93211164bef0900bd952e9f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 02:30:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ce4126c878c4ed29e38930c9d93b9fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582a7b5d786948f7a865872b4d1174b8 : SUCCESS in 5m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ccfc03427c4a440f89b19a64ab248a36 : SUCCESS in 7m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4000fb01a8ba41d599d798f65ae0e439 : SUCCESS in 6m 58s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7394e358d2da432797724c7085f33362 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c31101637035a49be2f0a1a4a86a44ca525e7c52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 02:30:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b654be6a5c11e6c1edb85ac698fed5217f20d242","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 02:31:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc62651bf3f4964a68b2ecee6a59a83\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997075d2d5a5417fbc7b68abb1408307 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a41e9f2825654a3e15ce3e5c4ded861193c6c4e1","revisions":{"b852ede3b28e2f606bf6b459c61262575b4e5d54":{"kind":"REWORK","_number":1,"created":"2024-09-26 06:06:17.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/34/930534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/34/930534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/34/930534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/34/930534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/34/930534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/34/930534/1"}}},"commit":{"parents":[{"commit":"a1044d4f8a8cbd8a579d793d3167c237bf080513","subject":"feat: update package for release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1044d4f8a8cbd8a579d793d3167c237bf080513"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 06:03:16.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 06:03:16.000000000","tz":480},"subject":"Fix container tag for 2023.2 on doc","message":"Fix container tag for 2023.2 on doc\n\nCurrently, the container tag for 99cloud/skyline in the documentation is\nuified as \"latest\" for the corresponding latest branch, which is\ninconsistent with the inconsistent with the version branches.\n\nUpdate 99cloud/skyline container tag from \"latest\" to \"2023.2\" in\ndoc/source/install/docker-install-ubuntu.rst to align with the branch.\n\nChange-Id: If2e93c1fa50f6745e7b670f01d212bd8872c2b5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b852ede3b28e2f606bf6b459c61262575b4e5d54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b852ede3b28e2f606bf6b459c61262575b4e5d54"}]},"branch":"refs/heads/stable/2023.2"},"a41e9f2825654a3e15ce3e5c4ded861193c6c4e1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-23 09:48:34.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"real_uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/34/930534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/34/930534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/34/930534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/34/930534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/34/930534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/34/930534/2"}}},"commit":{"parents":[{"commit":"adbbbbb98a33b6d972820cd27c946f2bc2549bef","subject":"Upgrade to use node 16 instead of node 14","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/adbbbbb98a33b6d972820cd27c946f2bc2549bef"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 06:03:16.000000000","tz":480},"committer":{"name":"HanGuangyu","email":"hanguangyu2@gmail.com","date":"2024-12-23 09:48:34.000000000","tz":0},"subject":"Fix container tag for 2023.2 on doc","message":"Fix container tag for 2023.2 on doc\n\nCurrently, the container tag for 99cloud/skyline in the documentation is\nuified as \"latest\" for the corresponding latest branch, which is\ninconsistent with the inconsistent with the version branches.\n\nUpdate 99cloud/skyline container tag from \"latest\" to \"2023.2\" in\ndoc/source/install/docker-install-ubuntu.rst to align with the branch.\n\nChange-Id: If2e93c1fa50f6745e7b670f01d212bd8872c2b5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a41e9f2825654a3e15ce3e5c4ded861193c6c4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a41e9f2825654a3e15ce3e5c4ded861193c6c4e1"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
