)]}'
{"id":"openstack%2Fskyline-console~930535","triplet_id":"openstack%2Fskyline-console~stable%2F2023.1~Ib1763e2e32b6224f19ffbdc24d7e3f31788d6dc4","project":"openstack/skyline-console","branch":"stable/2023.1","topic":"doc/align-container-tag-with-branch","attention_set":{},"removed_from_attention_set":{"33786":{"account":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"last_update":"2024-12-05 17:08:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib1763e2e32b6224f19ffbdc24d7e3f31788d6dc4","subject":"Fix container tag for 2023.1 on doc","status":"ABANDONED","created":"2024-09-26 06:28:51.000000000","updated":"2024-12-05 17:08:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24eb5a972c23581fc4f7af566cb2d8496b3fe372","_number":930535,"virtual_id_number":930535,"owner":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-26 10:37:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 06:47:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-05 17:08:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"d397f9d01e38a3178c33d3ad811014327038da69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 06:28:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d037c96cad50ee21f1ebbf0fbb1e896ded77f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 06:47:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57c9481106b14e859a0dbaca641660de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a74719dc9a743f58d4d1f3113609f06 : SUCCESS in 4m 34s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8deb89165d7e43c8846d3a6314e9a8b7 : RETRY_LIMIT in 2m 15s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2c7ee69f90e7418fb967717f5566037d : RETRY_LIMIT in 2m 05s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/dc27b9ce2912464ebc71e45c5390235e : RETRY_LIMIT in 2m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3b158b5543af601b4110ca79fe10053aa843b4af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 10:17:23.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ec6a71670e303db417d2554450edae69807198b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2024-09-26 10:18:00.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c7e8ad32ef4ddd82844a5f79a525602e765443de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 10:37:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44aa901feb7d40a6af5dee7169a2e620\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f7cc84adff24fa29502292574ebc23f : SUCCESS in 4m 01s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e2576544e6e74377be881f71b42ec325 : RETRY_LIMIT in 2m 04s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6a27131e7ef649138ed39c9e77ae8c62 : RETRY_LIMIT in 2m 21s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/83d0d808590c48d499f0a0f03649f619 : RETRY_LIMIT in 2m 25s","accounts_in_message":[],"_revision_number":3},{"id":"24eb5a972c23581fc4f7af566cb2d8496b3fe372","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-05 17:08:14.000000000","message":"Abandoned\n\nstable/2023.1 branch of openstack/skyline-console is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/2023.1 if you want to further work on this patch.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"acb284c391859a504f75d4a398948736f3d1a5a6","revisions":{"3cfd3d89ab3540c429bfef8d6e4945e79b99ba44":{"kind":"REWORK","_number":1,"created":"2024-09-26 06:28:51.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/35/930535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/35/930535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/1"}}},"commit":{"parents":[{"commit":"d4cae6dd6a96d74b754c590fd2dd48fab83a345f","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2023.1\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d4cae6dd6a96d74b754c590fd2dd48fab83a345f"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 06:09:19.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 06:09:19.000000000","tz":480},"subject":"Fix container tag for 2023.1 on doc","message":"Fix container tag for 2023.1 on doc\n\nCurrently, the container tag for 99cloud/skyline in the documentation is\nuified as \"latest\" for the corresponding latest branch, which is\ninconsistent with the inconsistent with the version branches.\n\nUpdate 99cloud/skyline container tag from \"latest\" to \"2023.1\" in\ndoc/source/install/docker-install-ubuntu.rst to align with the branch.\n\nChange-Id: Ib1763e2e32b6224f19ffbdc24d7e3f31788d6dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3cfd3d89ab3540c429bfef8d6e4945e79b99ba44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3cfd3d89ab3540c429bfef8d6e4945e79b99ba44"}]},"branch":"refs/heads/stable/2023.1"},"41948e5c4e8b31336f620e3498d9c1311ed6e5da":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-26 10:17:23.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/35/930535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/35/930535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/2"}}},"commit":{"parents":[{"commit":"d4cae6dd6a96d74b754c590fd2dd48fab83a345f","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2023.1\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d4cae6dd6a96d74b754c590fd2dd48fab83a345f"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 06:09:19.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 10:17:03.000000000","tz":480},"subject":"Fix container tag for 2023.1 on doc","message":"Fix container tag for 2023.1 on doc\n\nCurrently, the container tag for 99cloud/skyline in the documentation is\nuified as \"latest\" for the corresponding latest branch, which is\ninconsistent with the version branches.\n\nUpdate 99cloud/skyline container tag from \"latest\" to \"2023.1\" in\ndoc/source/install/docker-install-ubuntu.rst to align with the branch.\n\nChange-Id: Ib1763e2e32b6224f19ffbdc24d7e3f31788d6dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/41948e5c4e8b31336f620e3498d9c1311ed6e5da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/41948e5c4e8b31336f620e3498d9c1311ed6e5da"}]},"branch":"refs/heads/stable/2023.1"},"acb284c391859a504f75d4a398948736f3d1a5a6":{"kind":"NO_CHANGE","_number":3,"created":"2024-09-26 10:18:00.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/35/930535/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/35/930535/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/35/930535/3"}}},"commit":{"parents":[{"commit":"d4cae6dd6a96d74b754c590fd2dd48fab83a345f","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2023.1\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d4cae6dd6a96d74b754c590fd2dd48fab83a345f"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 06:09:19.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu2@gmail.com","date":"2024-09-26 10:17:50.000000000","tz":480},"subject":"Fix container tag for 2023.1 on doc","message":"Fix container tag for 2023.1 on doc\n\nCurrently, the container tag for 99cloud/skyline in the documentation is\nuified as \"latest\" for the corresponding latest branch, which is\ninconsistent with the version branches.\n\nUpdate 99cloud/skyline container tag from \"latest\" to \"2023.1\" in\ndoc/source/install/docker-install-ubuntu.rst to align with the branch.\n\nChange-Id: Ib1763e2e32b6224f19ffbdc24d7e3f31788d6dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/acb284c391859a504f75d4a398948736f3d1a5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/acb284c391859a504f75d4a398948736f3d1a5a6"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
