)]}'
{"id":"openstack%2Fskyline-console~930864","triplet_id":"openstack%2Fskyline-console~stable%2F2024.2~If941e34fbd82d9a53cd113ee8fa0e74d69a12f29","project":"openstack/skyline-console","branch":"stable/2024.2","topic":"2024-2-release","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-10-26 15:09:21.000000000","reason":"Change was submitted"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2024-10-26 14:04:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}},"hashtags":[],"change_id":"If941e34fbd82d9a53cd113ee8fa0e74d69a12f29","subject":"feat: update package for release","status":"MERGED","created":"2024-09-30 06:08:20.000000000","updated":"2024-10-26 15:10:21.000000000","submitted":"2024-10-26 15:09:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930864-2024-2-release","meta_rev_id":"a2fe2369176ed46a28e2dfe6c0d823150883781d","_number":930864,"virtual_id_number":930864,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-26 15:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-10-26 14:04:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-10-26 14:04:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-30 07:18:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-30 09:16:16.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a76816b01da3af0ac230f495cfa19d3aa8c386a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-09-30 06:08:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d89f7a54ed228816be0e2c6498d619bdb34b446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 07:18:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5aeb7a111745938c81cb3a7771ba4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53863d8a39ae4b8abddfd7c6c52cfe01 : SUCCESS in 3m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e8a577f3d78d4603aa0ac8ad14cace05 : SUCCESS in 4m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6313979095b54ff2a5672023573bd9e0 : SUCCESS in 5m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/825d8ac3a3ec4401b7999143cf6e0688 : SUCCESS in 6m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e74d4e51f8944e02a0d046d561268408 : SUCCESS in 1h 09m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6257a4d3ae9c445eb103a2cd8e275715 : SUCCESS in 56m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/42fdcc79d58d4819b7aac1054f212b12 : SUCCESS in 59m 58s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/97aa3835d0a44543b2230e1c010d6288 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"36de1e537c96914bdafedaff168fbf1884fbfd18","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-10-26 14:04:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32eb37f63f3b1be1e100bce24be1de580bfabfcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 14:05:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a412902d164666a0b1b8fa65148a5a32d3b8e879","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 15:09:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92ec0adbca574804b281c14a36b9076c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ef1c6c4a8c4673ac52e4dffa750760 : SUCCESS in 4m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/076a48db143e4cc4b1cc626d5a4447e0 : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/86256b189a4843c5a291d9b88ae2ad13 : SUCCESS in 4m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6ab02c5cc0724138935503b09877e738 : SUCCESS in 5m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/75ebe2b825fd43538fe194f4f757a838 : SUCCESS in 56m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/781fa25e796742e9965e384b1f7a51f6 : SUCCESS in 57m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3fb77c3b25bb4272835db20bd6e1a585 : SUCCESS in 52m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7369caa91e0f428fbfa8b8a8ab1f1635 : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fe63b87a5ca8f5dbd2f8a57ea4d431a1c809280c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 15:09:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a2fe2369176ed46a28e2dfe6c0d823150883781d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 15:10:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/578656cbff9b49eb90b2ecad2563a70d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1782d6b860164648a6210d3fce09a67e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44415b833314776d92b0767660fd5595c99bfd1a","revisions":{"44415b833314776d92b0767660fd5595c99bfd1a":{"kind":"REWORK","_number":1,"created":"2024-09-30 06:08:20.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/64/930864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/930864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/930864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/930864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/930864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/930864/1"}}},"commit":{"parents":[{"commit":"5501bad74a522df44e2fbaa67796856782c70dcd","subject":"feat: add Turkish language to the default config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5501bad74a522df44e2fbaa67796856782c70dcd"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-09-30 06:06:40.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-09-30 06:06:40.000000000","tz":480},"subject":"feat: update package for release","message":"feat: update package for release\n\n1. update package for branch 2024.2 release\n2. update gitreview file\u0027s defaultbranch\n\nChange-Id: If941e34fbd82d9a53cd113ee8fa0e74d69a12f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/44415b833314776d92b0767660fd5595c99bfd1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/44415b833314776d92b0767660fd5595c99bfd1a"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
