)]}'
{"id":"openstack%2Fskyline-console~935601","triplet_id":"openstack%2Fskyline-console~master~I338c2f60c1fec148613ecbd5d183d2e4cdd049c6","project":"openstack/skyline-console","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-01-17 05:03:27.000000000","reason":"Change was submitted"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2025-01-17 04:50:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-01-17 05:03:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I338c2f60c1fec148613ecbd5d183d2e4cdd049c6","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","status":"MERGED","created":"2024-11-19 03:35:15.000000000","updated":"2025-01-17 05:05:24.000000000","submitted":"2025-01-17 05:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935601-migrate-to-noble","meta_rev_id":"9288b360098782ef680e00405905a9d5953aad52","_number":935601,"virtual_id_number":935601,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-17 05:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2025-01-17 04:50:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2025-01-17 04:50:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 03:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-05 20:50:02.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2024-12-06 17:52:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"c8cebd2783bfdd0fe9760d79a6a9c84b93c0785b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 03:35:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e908f47a6a8b061a01d1fd892782b5111d7e1cb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 03:48:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1483c4b4f5c64e8fb7a9336a5c26130a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf580baa6266483b947a41d58c315f8e : SUCCESS in 4m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7d25d24b7c27432ebd08b2f8448c7823 : SUCCESS in 9m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8f3154a5ffe449649979b33b27d61860 : SUCCESS in 4m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cf40c289b3e84c2bb14b03b22b074e9c : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f8577714ba727c60a6ce9e6d2a7d57793b08ef5f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-01-17 04:50:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d8c0a262f542233d14e56678287b2405f87f5cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 04:50:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a236a33c16a55cbaa43cd0785825af079bf3766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 05:03:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0087efbf77fc4ad3bb5c5835672e3572\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7272a50fe3e343808a5b6be680a9f415 : SUCCESS in 3m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/49302330d2614dac93e6bab5a2819614 : SUCCESS in 8m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6c08ab0a623f4ee1b82e3a25e9de1620 : SUCCESS in 7m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0eca996a480749769654d0d776a53d39 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0bb4c6be5039f6aa5db3b55348ce7c86c2e3bb5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 05:03:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9288b360098782ef680e00405905a9d5953aad52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 05:05:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff20e578b0804061ac1207cc98978434\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/191c2dc81da342dea5b18392e2336e47 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b751bc3c5017fa7d8fcfc32544f8e31c4051552","revisions":{"9b751bc3c5017fa7d8fcfc32544f8e31c4051552":{"kind":"REWORK","_number":1,"created":"2024-11-19 03:35:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/935601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/01/935601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/935601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/935601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/935601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/01/935601/1"}}},"commit":{"parents":[{"commit":"b40736c9e95e8316309029b7bdb4a1d91c3a1b65","subject":"Merge \"Update master for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b40736c9e95e8316309029b7bdb4a1d91c3a1b65"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 03:33:58.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 03:35:13.000000000","tz":-480},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble), updating the jobs nodeset to noble.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I338c2f60c1fec148613ecbd5d183d2e4cdd049c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9b751bc3c5017fa7d8fcfc32544f8e31c4051552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9b751bc3c5017fa7d8fcfc32544f8e31c4051552"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
