)]}'
{"id":"openstack%2Fskyline-console~938174","triplet_id":"openstack%2Fskyline-console~stable%2F2023.2~I81411ec4824f19b1eaf03c2b1c469f6bfb16d82a","project":"openstack/skyline-console","branch":"stable/2023.2","hashtags":[],"change_id":"I81411ec4824f19b1eaf03c2b1c469f6bfb16d82a","subject":"Upgrade to use node 16 instead of node 14","status":"MERGED","created":"2024-12-23 07:34:30.000000000","updated":"2024-12-23 08:29:36.000000000","submitted":"2024-12-23 08:27:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938174","cherry_pick_of_change":928111,"cherry_pick_of_patch_set":2,"meta_rev_id":"acdcdbce761b4219dee17320e470a2d1e0908dad","_number":938174,"virtual_id_number":938174,"owner":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-23 08:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-12-23 08:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-12-23 08:21:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-23 07:47:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e2822de928b747aece16313565783e315773187","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-12-23 07:34:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6ce5db200ab3cb8c88e94e6dda8839199e9fdb3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 07:47:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60b4869b338547718a08cbe4ba0e0cdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837fc42d28384edaab6ea3576bb07354 : SUCCESS in 3m 33s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ffeada999a4140af9526ebf1be699e34 : SUCCESS in 6m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8a7239f101414ed29f2e12d238e69c90 : SUCCESS in 6m 53s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/83fef56f5054477198dffd30e4e43b7a : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"df20ca03c3b12df46b0c347db0005c2b9a71d41a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-12-23 08:21:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34a6ac825a093c709a72b2104f9cd3bddc2e6cf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 08:22:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81284f2b07e0feb5d19e7ce1794862035f230b1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 08:27:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d516eb6cba9845489c8c459cfe14751f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9650861e3b6415eb06ffe1cbfed3f28 : SUCCESS in 3m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7ee1fc2b81884cf280cd44fd5dee3919 : SUCCESS in 5m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a0ee3ef9da614fd1a60b9bfa42c78e64 : SUCCESS in 4m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/01e790fb705d42ca87606f8354fc2648 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f2bee412d63e5426629b859e1f60fab10b03c855","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 08:27:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"acdcdbce761b4219dee17320e470a2d1e0908dad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 08:29:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae5210aea6447faaeba94f297da51da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b260895519cf49f899010a8d9e08b615 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adbbbbb98a33b6d972820cd27c946f2bc2549bef","revisions":{"adbbbbb98a33b6d972820cd27c946f2bc2549bef":{"kind":"REWORK","_number":1,"created":"2024-12-23 07:34:30.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/74/938174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/74/938174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/938174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/938174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/938174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/74/938174/1"}}},"commit":{"parents":[{"commit":"a1044d4f8a8cbd8a579d793d3167c237bf080513","subject":"feat: update package for release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1044d4f8a8cbd8a579d793d3167c237bf080513"}]}],"author":{"name":"zhuboxiang","email":"zhu.boxiang@99cloud.net","date":"2024-09-05 02:16:33.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2024-12-23 07:34:30.000000000","tz":0},"subject":"Upgrade to use node 16 instead of node 14","message":"Upgrade to use node 16 instead of node 14\n\nFailed to update apt cache: W:The repository\n\u0027https://deb.nodesource.com/node_14.x noble Release\u0027\ndoes not have a Release file.\n\nSo Upgrade to use node 16 instead of node 14\n\nChange-Id: I81411ec4824f19b1eaf03c2b1c469f6bfb16d82a\n(cherry picked from commit 6b85304c01834f96992110100aa0574e569dc55c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/adbbbbb98a33b6d972820cd27c946f2bc2549bef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/adbbbbb98a33b6d972820cd27c946f2bc2549bef"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
