)]}'
{"id":"openstack%2Fskyline-console~939582","triplet_id":"openstack%2Fskyline-console~master~Id556e0e53d35eb430fece6267977b062b471c654","project":"openstack/skyline-console","branch":"master","topic":"rhel-install","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-01-19 10:35:11.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-02-11 05:33:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id556e0e53d35eb430fece6267977b062b471c654","subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","status":"MERGED","created":"2025-01-18 12:30:51.000000000","updated":"2025-02-11 05:35:10.000000000","submitted":"2025-02-11 05:33:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939582-rhel-install","meta_rev_id":"c784ee60d3479f64a808cf0651afb811f9899938","_number":939582,"virtual_id_number":939582,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 05:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-01-19 10:35:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-02-11 05:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-02-11 05:18:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-18 13:03:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-18 22:12:38.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-02-11 05:18:27.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"2ba0ab20ac6a8cffe72fc6824f98321e973352ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 12:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c34d716dabd05d0a7065843336b117a0c275d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 13:03:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff51d72fedbc4a479aa9d12f35cb813a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dda07faefa08415c9e72b79ab8178a9f : SUCCESS in 2m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2679ec966b74476b8807e8070161704 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1afe3240d344d3a841f70e0ad5f0c24 : SUCCESS in 4m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7211814bd1a9476ead7eb88be05ec9ed : SUCCESS in 6m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7bf16553b37b460391c7369b9d561de0 : SUCCESS in 4m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/64dab435ad1444dc935e11b33624905b : SUCCESS in 4m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7eff66f1b65041c68760ecbc1d58e2ca : FAILURE in 20m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3e1b79bb4ebb40ecabc069572ad0d909 : FAILURE in 12m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f39cad49fd7c427da5cc2c43bc6877be : FAILURE in 15m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d079a4a70b854ae88c8cbc26b7bfae31 : FAILURE in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0a09fb644b7eb2ff896a75ca9eb0313ee30fb859","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 14:01:38.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bfd4798bfaa298e7bb6752e10c6a16c58f0cb4b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 14:02:39.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"df80874ba60cb1151c3cfa49c0a76a171347b26f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 14:07:17.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"235e837f2ef7c25d3676b947f196b3d434da9918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 14:24:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0ee6d0e727b4df0b96531f88d2df381\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1527cce233e64db496435a7035593f31 : FAILURE in 3m 27s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8094b8b2e00047318f56a106dcb50087 : SUCCESS in 10m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f8a636de62ac471ba6e6837bc3be8d74 : SUCCESS in 7m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4c28f3e68f404409b64242dc5e09af88 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":4},{"id":"5d498baa6517a77f083e10af71c93d816fdf83fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 14:32:47.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7f95769fea65d0b292d5d89d266516d3b98eada5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 14:33:15.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"2addeacde7303cb3bdbd2b9e25be10b0628ee219","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 14:34:05.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"07505a32e2a7d1bbf4605509efa9c313bc880872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 14:52:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaaa8c747927453fba2c2d2d869aa4cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d357e50d564339977f7bcb88a6f8c2 : FAILURE in 3m 33s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d68ce1ee5535446ca9265c39f6745d10 : SUCCESS in 11m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5f475ce3352d472dad11bc8ed4c0c765 : SUCCESS in 6m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d558a9d8ed3a47b0ba1f4e532e2febb9 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":7},{"id":"fc61801f61fa0ab6355c32b26e5b462e439a2f85","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 21:30:53.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"117032879a1328615170d85f524029c1dde356de","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 21:30:57.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"cd740133b23164cc963ea717df09df49ab0152ae","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 21:31:12.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":8},{"id":"d542ebb7fd0805b0e6f68404659c26b75ae18728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 21:47:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f26f4230bbde492da1f95d6879ec038d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdef8e228b524b4db71554d94f8a70b7 : SUCCESS in 3m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/644b4763c6324423abd6c8833813bffc : SUCCESS in 10m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/be57766bf7a045e6b21174a9e2bcc938 : SUCCESS in 7m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/da5eb45365eb4c97a468bce3981afc6f : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":8},{"id":"02e495f2a64498e241b28eb11c323492f9ad7aab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 21:55:51.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"509df0085e2a69e09396782de1cd72890c66ea55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 21:55:59.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"008b9a2df41b759a290363b0fb2ca51b8ccc7d67","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 21:56:11.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":10},{"id":"0012c3bcaf73e0bf2d101be194577798d576fe49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 22:11:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9bd7340926b4bd2bd370ecc7c6de3a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580201953e27438aabb5169287510c11 : SUCCESS in 5m 00s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4cad24b6d8f74b1a915036992ea7e1f0 : SUCCESS in 10m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1cf58741a13549979aee5b1698840c50 : SUCCESS in 7m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/25bf013b7b194d068b116997806519f5 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":10},{"id":"2ad9d0bc987558f4fff4ec992d6b471c136fbbe5","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-01-18 22:12:38.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"5a8cc68458cfab33f43f1cbe86ecbc9130178bcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 23:44:09.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"fc82488be4c0444f23e069fe6be04779874011ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-18 23:44:18.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"f9619f65b88e7fc520e7f294156f47a18e48bd20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-19 00:14:06.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/084bb40bf5ed4504bc827055bb2c3291\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c292faf010124488aa35f9349fdbf65e : FAILURE in 6m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8844e58d85c24877963e6c3e27e2796f : SUCCESS in 8m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fe42597d1f77446cad00ec7be28e37fb : SUCCESS in 7m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d0abc86b53b342a98dbc0f09ed842e8d : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":12},{"id":"0c8d39eedb2ff4391aedff4878291be5add29cdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-19 00:20:45.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"0a0703dc45f471c13088a91b7f2fd1113c7c20b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-19 00:27:33.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/991b9117c35245bcbabc84c7c14dbfa8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05e1c5e86c243628a4c38973da740d1 : FAILURE in 3m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6c76b86a891d4e4e9996948f78a8d1df : SUCCESS in 6m 05s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9ab8ba67c8d648708becaef474aa59d8 : SUCCESS in 4m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/716aba9cf74a439a92464796006dab73 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":13},{"id":"93a2006cd82a518573ae4245e698564b323d36bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-19 00:28:10.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"8cf570a54d3ce9f7e7ba18d5145d17fd9a5eb6f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-19 00:28:34.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"af3bb8686f937e024efb4e2c290fe74a2ac75400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-19 00:35:04.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74707b42a9c04b2eb958b1a19182cd29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e23eca4bff84defad9a691d736890e2 : FAILURE in 3m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8d47083b32ee45828c2a657ab767bff8 : SUCCESS in 6m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2f6dbd53062040a8883206b84f6186a9 : SUCCESS in 5m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5efd76781dea48c2af37a3767c04c93f : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":15},{"id":"caafed8a6a7474e3b6561a43e24e20c3eb14ceab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-19 07:47:28.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"4db6eee5d4ff2aeea2873f234780ec496591d49b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-19 08:32:16.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"c8f03f22f8c116ce6a4ebd84e33db79190e5d59e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-19 08:58:31.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de7cecb8e9da459fbc8e881c3afd7934\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b00cce16daa4ca88ae4b8fc26ab4ca3 : FAILURE in 3m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/203aae89800543049faac53b05d05c27 : SUCCESS in 11m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ce7f3c2f4cfa4dc0b79a52a79d1b90cc : SUCCESS in 7m 37s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8bc69d6e3b73493989aa43928b6dce0b : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":17},{"id":"a0e2d6be2936eb9014c8f5dcf3f483787b26ab88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-19 09:08:35.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"7f3cbbb9a14b473f327ee2a2dbea4797903157b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-19 09:21:26.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2625c8e23c24f22aa8eb8d435e9867e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f4e3c35e9d4c68b02009c2baea9957 : FAILURE in 5m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a85ea22a95aa44998891154c0747405f : SUCCESS in 11m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/94cf51a4b9b748b5b343cff80bbe7ca4 : SUCCESS in 5m 04s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/10b9be97226345139a698f4207b163c0 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":18},{"id":"ca80de4ad26ef475cd159ba83d9abf05b392fd92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-19 09:26:45.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"f799f403afc2c1d5e88a49c38da4a8802c2f0a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-19 09:32:31.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da70f830c6a6475b81b6947239780fa2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485449a9581847809f1fb062d50e4472 : SUCCESS in 3m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/29a1728e9e0842b984d701a5db1a50b8 : SUCCESS in 4m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7faf66f4888d43d1bf123253aef57ae3 : SUCCESS in 3m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fa96f98770ab407e9d0869455c16498b : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":19},{"id":"a51298143f5e4c3837066df4c2bb21c3476ed644","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-01-19 10:35:11.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"3743db9dae7ead7e8f8e3055ca150a98ecd52e78","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-11 05:18:27.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"078413460763210a78d96d12ef4f65680f53d1b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 05:18:46.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"0f384d7044efec54b64c9e0facc2f9dd7470b151","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 05:33:20.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c7c99bcfcd444d8ff82e70aaae7989\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e78f381dc3f42ad88b6371f4f768c57 : SUCCESS in 4m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/125859829b2f4382a88860759df9795e : SUCCESS in 6m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e0b8cdc9f0d84ee78a01c5915d0f8d17 : SUCCESS in 5m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/587645f0a485400b95e642b0a8cc5dfc : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":19},{"id":"7a15870ebaf707594fe95bf804b116f402f72f42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 05:33:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"c784ee60d3479f64a808cf0651afb811f9899938","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 05:35:10.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf089be730e410b80271c6bcaba15ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f263ab272434bfca67308bb71f39b04 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"2f59370762ddcfd28ae5f21830cf14ad3f5ac900","revisions":{"b8c6c2ef6b321484cd77fc350a3dc4861ad164db":{"kind":"REWORK","_number":1,"created":"2025-01-18 12:30:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/1"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"subject":"Install from Source - RHEL","message":"Install from Source - RHEL\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b8c6c2ef6b321484cd77fc350a3dc4861ad164db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b8c6c2ef6b321484cd77fc350a3dc4861ad164db"}]},"branch":"refs/heads/master"},"e24875fb769e2974fdf6e2ead37e72e12d535f68":{"kind":"REWORK","_number":2,"created":"2025-01-18 14:01:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/2"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 14:01:29.000000000","tz":0},"subject":"Install from Source - RHEL","message":"Install from Source - RHEL\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e24875fb769e2974fdf6e2ead37e72e12d535f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e24875fb769e2974fdf6e2ead37e72e12d535f68"}]},"branch":"refs/heads/master"},"a4d3b98a67b4074d67793fcd1841654c39c49472":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-01-18 14:02:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/3"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 14:02:39.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL9","message":"Skyline Installation Guide - for RHEL9\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a4d3b98a67b4074d67793fcd1841654c39c49472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a4d3b98a67b4074d67793fcd1841654c39c49472"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6c09e4da378fa19991ebf8e0da708776cd3a2a51":{"kind":"REWORK","_number":4,"created":"2025-01-18 14:07:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/4"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 14:05:15.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL9","message":"Skyline Installation Guide - for RHEL9\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6c09e4da378fa19991ebf8e0da708776cd3a2a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6c09e4da378fa19991ebf8e0da708776cd3a2a51"}]},"branch":"refs/heads/master"},"1510d73277ebc17ddd5c19b9428d16c7f43f60cc":{"kind":"REWORK","_number":5,"created":"2025-01-18 14:32:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/5"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 14:32:38.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL9","message":"Skyline Installation Guide - for RHEL9\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1510d73277ebc17ddd5c19b9428d16c7f43f60cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1510d73277ebc17ddd5c19b9428d16c7f43f60cc"}]},"branch":"refs/heads/master"},"f4c762de9ed856c3290e1d33c30ff0343c291872":{"kind":"REWORK","_number":6,"created":"2025-01-18 14:33:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/6"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 14:33:09.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL9","message":"Skyline Installation Guide - for RHEL9\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4c762de9ed856c3290e1d33c30ff0343c291872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4c762de9ed856c3290e1d33c30ff0343c291872"}]},"branch":"refs/heads/master"},"716840fcc01360fc95ca61b8e1a148e828cafc92":{"kind":"REWORK","_number":7,"created":"2025-01-18 14:34:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/7"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 14:34:05.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL9","message":"Skyline Installation Guide - for RHEL9\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/716840fcc01360fc95ca61b8e1a148e828cafc92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/716840fcc01360fc95ca61b8e1a148e828cafc92"}]},"branch":"refs/heads/master"},"2c96380ddeb6d86f5a52d38199cc84885554a3c8":{"kind":"REWORK","_number":8,"created":"2025-01-18 21:30:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/8"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 21:30:53.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL9","message":"Skyline Installation Guide - for RHEL9\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2c96380ddeb6d86f5a52d38199cc84885554a3c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2c96380ddeb6d86f5a52d38199cc84885554a3c8"}]},"branch":"refs/heads/master"},"f3ffc01f75719833f57a4f870a0598b3388ab19d":{"kind":"REWORK","_number":9,"created":"2025-01-18 21:55:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/9"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 21:55:41.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL9","message":"Skyline Installation Guide - for RHEL9\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3ffc01f75719833f57a4f870a0598b3388ab19d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3ffc01f75719833f57a4f870a0598b3388ab19d"}]},"branch":"refs/heads/master"},"24fb1c70c9e042ee4b5cd9dce2abbb18b33d68ef":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2025-01-18 21:55:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/10"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 21:55:59.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS","message":"Skyline Installation Guide - for RHEL/CentOS\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/24fb1c70c9e042ee4b5cd9dce2abbb18b33d68ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/24fb1c70c9e042ee4b5cd9dce2abbb18b33d68ef"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0e96cb5485b94872344a57f1affa242ac0992870":{"kind":"REWORK","_number":11,"created":"2025-01-18 23:44:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/11"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 23:44:09.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS","message":"Skyline Installation Guide - for RHEL/CentOS\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0e96cb5485b94872344a57f1affa242ac0992870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0e96cb5485b94872344a57f1affa242ac0992870"}]},"branch":"refs/heads/master"},"5896adaec34f26597a63fba46420cf9c08020594":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2025-01-18 23:44:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/12"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 23:44:18.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","message":"Skyline Installation Guide - for RHEL/CentOS - Docker\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5896adaec34f26597a63fba46420cf9c08020594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5896adaec34f26597a63fba46420cf9c08020594"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2bad33bf8d637f5b2b3b9f5eeb30c792369ab2b8":{"kind":"REWORK","_number":13,"created":"2025-01-19 00:20:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/13"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-19 00:20:45.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","message":"Skyline Installation Guide - for RHEL/CentOS - Docker\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2bad33bf8d637f5b2b3b9f5eeb30c792369ab2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2bad33bf8d637f5b2b3b9f5eeb30c792369ab2b8"}]},"branch":"refs/heads/master"},"a68d442206f39899adb0a15c13d5deb45ef590b4":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-01-19 00:28:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/14"}}},"commit":{"parents":[{"commit":"629ef92824bec54247cf6bea9b8c8d278efc336b","subject":"Skyline Installation Guide - for RHEL/CentOS - Source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/629ef92824bec54247cf6bea9b8c8d278efc336b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-19 00:28:10.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","message":"Skyline Installation Guide - for RHEL/CentOS - Docker\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a68d442206f39899adb0a15c13d5deb45ef590b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a68d442206f39899adb0a15c13d5deb45ef590b4"}]},"branch":"refs/heads/master","description":"Rebase"},"130a50664abd3e87b7fc6d99bd6145f76f7eabfe":{"kind":"REWORK","_number":15,"created":"2025-01-19 00:28:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/15"}}},"commit":{"parents":[{"commit":"629ef92824bec54247cf6bea9b8c8d278efc336b","subject":"Skyline Installation Guide - for RHEL/CentOS - Source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/629ef92824bec54247cf6bea9b8c8d278efc336b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-19 00:28:34.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","message":"Skyline Installation Guide - for RHEL/CentOS - Docker\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/130a50664abd3e87b7fc6d99bd6145f76f7eabfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/130a50664abd3e87b7fc6d99bd6145f76f7eabfe"}]},"branch":"refs/heads/master"},"6d8e3853881748bf525031822fff24c970568027":{"kind":"REWORK","_number":16,"created":"2025-01-19 07:47:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/16"}}},"commit":{"parents":[{"commit":"629ef92824bec54247cf6bea9b8c8d278efc336b","subject":"Skyline Installation Guide - for RHEL/CentOS - Source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/629ef92824bec54247cf6bea9b8c8d278efc336b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-19 07:47:27.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","message":"Skyline Installation Guide - for RHEL/CentOS - Docker\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6d8e3853881748bf525031822fff24c970568027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6d8e3853881748bf525031822fff24c970568027"}]},"branch":"refs/heads/master"},"3f8d17027405a238af4ea0c0024a1c13404ca208":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2025-01-19 08:32:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/17"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-19 08:32:16.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","message":"Skyline Installation Guide - for RHEL/CentOS - Docker\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f8d17027405a238af4ea0c0024a1c13404ca208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f8d17027405a238af4ea0c0024a1c13404ca208"}]},"branch":"refs/heads/master","description":"Rebase"},"752dfe6a312a5d90ac29760b0ca080aba5eec3cf":{"kind":"REWORK","_number":18,"created":"2025-01-19 09:08:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/18"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-19 09:08:35.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","message":"Skyline Installation Guide - for RHEL/CentOS - Docker\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/752dfe6a312a5d90ac29760b0ca080aba5eec3cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/752dfe6a312a5d90ac29760b0ca080aba5eec3cf"}]},"branch":"refs/heads/master"},"2f59370762ddcfd28ae5f21830cf14ad3f5ac900":{"kind":"REWORK","_number":19,"created":"2025-01-19 09:26:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/939582/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/939582/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/939582/19"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-18 12:30:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-01-19 09:26:45.000000000","tz":0},"subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","message":"Skyline Installation Guide - for RHEL/CentOS - Docker\n\nChange-Id: Id556e0e53d35eb430fece6267977b062b471c654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2f59370762ddcfd28ae5f21830cf14ad3f5ac900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2f59370762ddcfd28ae5f21830cf14ad3f5ac900"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
