)]}'
{"id":"openstack%2Fskyline-console~939685","triplet_id":"openstack%2Fskyline-console~master~I18b88e6f8ab5f5e15c4bb0dc4292cb82f6a8107e","project":"openstack/skyline-console","branch":"master","topic":"fix-ci","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2025-02-12 21:05:31.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-02-12 21:05:31.000000000","reason":"Change was abandoned"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-01-24 15:16:39.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I18b88e6f8ab5f5e15c4bb0dc4292cb82f6a8107e","subject":"Fix Makefile for newer Ubuntu","status":"ABANDONED","created":"2025-01-21 07:55:50.000000000","updated":"2025-02-12 21:05:31.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3e501e19c106ee2a16f5b53b2f146c186cbc45d1","_number":939685,"virtual_id_number":939685,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-24 17:36:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2025-02-11 17:14:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-21 08:16:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-21 20:36:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-01-23 11:33:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-01-23 13:38:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2025-02-11 05:40:22.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"CC"},{"updated":"2025-02-11 17:14:35.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2025-02-12 12:16:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"}],"messages":[{"id":"dd6ab09a3e06ef6d8fd3efef4b22bf737775afd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-21 07:55:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80ef5d3ed81bd47a652b11e69c48c22181eec3c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 08:16:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf730f77feb44879b1bb7dad7a066a1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3139dd131e944038b8bfd7dc48df9996 : SUCCESS in 4m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a3e7c9e946584918a878fdd3389e4891 : SUCCESS in 7m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8dd099f5b97f45ea866d4c65d60f39a4 : SUCCESS in 6m 01s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/33989afce0e542fd80891ca79b9ac6c8 : SUCCESS in 3m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/57fee7c0e62a4b28b2da878907b43fd1 : FAILURE in 19m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c0bf97a7044745f2a13de634b1c90088 : FAILURE in 13m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2d294d0e69d84ca1a2657d18dcba5c7e : FAILURE in 17m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4caa8597820b4577a35d8b87c475e9ce : FAILURE in 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"db4766b22713c1569f35512502753114d1e569df","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-21 20:36:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dfd4e01b9c2342c31ab490474837238549c36daf","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-23 09:58:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93b95901f2055a0ae695f6cddc9dfb3a707f0ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-23 10:00:12.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3f6efa5d6d31135062ab6fc47f494a31837cb266","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-23 11:33:19.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ee5701b0813e2bec3cd87026586380d517925df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 12:49:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29ca07006fb0427bbd8754b3f5cab37c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78810aadbc69443298001c6bfb15d70b : SUCCESS in 4m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c232c9a4192042128775b3d78ca08bef : SUCCESS in 10m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e8501ae5aa274781a8d9e800f963d39f : SUCCESS in 8m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b5d5189e92bc45d9ab58ac0b7b806d6f : SUCCESS in 8m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d4f0454506ec4f51a6a1502fc4c6e68b : FAILURE in 2h 16m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0dff885039e8467b8053fc1a28d341e9 : FAILURE in 1h 05m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/14c525531f1e4436bff35fe32db355a2 : FAILURE in 44m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3241603abd5940ec9734cba1ab4dea1b : FAILURE in 1h 22m 03s","accounts_in_message":[],"_revision_number":2},{"id":"353882999050f0e3897ccdb06d542339c0c19768","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-23 13:38:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c404967f23b81ee54cbad0fd1bc2075e4ab098a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-24 14:49:51.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a35ed151b9760b6e4f9d930709874eec46c27cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 17:36:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/299aad2c06c943bd84b4b2df1664f6ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e24bc01cea4b908d0452a8f4ee5714 : SUCCESS in 3m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d386c59aa9374a52b8ed147b90031450 : SUCCESS in 11m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/989266d784cb4ed7b68edd4b9caf4f93 : SUCCESS in 4m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/91ee002772b64cf6b1b2f5de33760156 : SUCCESS in 7m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/29027ac085fc429fa3c017b0b43b4b39 : FAILURE in 2h 38m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c4c71846c79648ff96b7c3935d5f289d : FAILURE in 1h 41m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/10028acb63da451cb6b6f7ebddd21a8d : FAILURE in 1h 07m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1870116809c840f0b7709cdbb17584fe : FAILURE in 1h 48m 42s","accounts_in_message":[],"_revision_number":3},{"id":"068a2e5bb49cc17de579b57a9001ff02fccc37ef","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-11 05:40:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f2418c15c63b1b1a077464b9ee18a6a21ba2fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 07:33:11.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c37d2e8ce56d46c1ba05d81de73060e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ddccd7e6cb489bb563c9e387e5f77a : SUCCESS in 3m 31s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/94d3aed191854d66a1b2a9c35c7096ce : SUCCESS in 5m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8af348244f5a42ecbac891433693c752 : SUCCESS in 3m 37s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6dc79a005cbe4eb99c8fdb008a07c240 : SUCCESS in 4m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/effe2690e4074e539ea7bf243011e575 : FAILURE in 1h 33m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e1d09d8bc6034ac581fbb5c69fb745df : FAILURE in 1h 52m 09s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b85d25b1891c427aa40ff3691f906130 : FAILURE in 38m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8d87e89ae110409283702257d05723ac : FAILURE in 1h 17m 56s","accounts_in_message":[],"_revision_number":3},{"id":"5a6c9315d9fc9ee6e038012320a1dfff5302d0cd","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-11 17:14:35.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69cbe960342b797f708310ee812518baae5668e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 18:27:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16772e48ceda4055b955bdafd0c000ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a13bd9cc1c742efa42db89a2840488e : SUCCESS in 4m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2ecf0b6ba1d84f69a3e965e75f7172af : SUCCESS in 7m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/53bac639efd84c16b06c9bc3068aa8b0 : SUCCESS in 5m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9a58faa7150d439b876a78330a4b03c9 : SUCCESS in 5m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7125698ee2ab4f688f996f33bc2ca3bc : SUCCESS in 1h 10m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5f8edfa3498c43c8a562996e42d0e3cb : SUCCESS in 49m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/58ac3a43eb5940bda98c26ba32119b79 : FAILURE in 56m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/dbaed003273549c3943eb2808d04c168 : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":3},{"id":"9dc4e372b93b23863e3b6c31e636f30df36d4589","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-11 23:10:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1461ed684ef9c34a9f1f6b7614381dd990b4f370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 01:02:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d8697079eac41aaa7d6a95acdb749c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360771360e7242779614d37fb6e5557d : SUCCESS in 4m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/eb40ad0f53d7454d8001d7b3b0075d8d : SUCCESS in 7m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/41b21954de89440581de72e45b322aa8 : SUCCESS in 5m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/148edf347db9489bb2ebd53104056d8d : SUCCESS in 7m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7299ac9790a44c87a4083573c4e61243 : SUCCESS in 1h 31m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6de5aecd724e496eaa69ae1dffb4b16d : SUCCESS in 46m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1bc3cbfd23de4be98fe276d38d01b36f : FAILURE in 49m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a0a61bed43bd4baca48495ec1eac0a37 : SUCCESS in 49m 22s","accounts_in_message":[],"_revision_number":3},{"id":"90a981f86d21a957c2b619e10253fd1b63ebb73c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-12 12:16:23.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_37598\u003e.","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":3},{"id":"3e501e19c106ee2a16f5b53b2f146c186cbc45d1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-12 21:05:31.000000000","message":"Abandoned\n\nintegrated into https://review.opendev.org/c/openstack/skyline-console/+/941303","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cbe81476ef2d7fe164369c515b90e257bac7761a","revisions":{"29a230411c6d75bdbc84f2db87b1b1d0a74f4702":{"kind":"REWORK","_number":1,"created":"2025-01-21 07:55:50.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/939685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/939685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/1"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-21 07:53:43.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-21 07:55:33.000000000","tz":60},"subject":"Fix Makefile for newer Ubuntu","message":"Fix Makefile for newer Ubuntu\n\n- Pinning setuptools seems no longer to be needed, also breaks due to\n  trying to do a global install\n- Use python3 instead of python binary\n\nChange-Id: I18b88e6f8ab5f5e15c4bb0dc4292cb82f6a8107e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/29a230411c6d75bdbc84f2db87b1b1d0a74f4702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/29a230411c6d75bdbc84f2db87b1b1d0a74f4702"}]},"branch":"refs/heads/master"},"57544775fa1c1e99f9dd157d9175f010ddb15c0c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-23 10:00:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/939685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/939685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/2"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-21 07:53:43.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-23 10:00:12.000000000","tz":0},"subject":"Fix Makefile for newer Ubuntu","message":"Fix Makefile for newer Ubuntu\n\n- Pinning setuptools seems no longer to be needed, also breaks due to\n  trying to do a global install\n- Use python3 instead of python binary\n\nDepends-On: https://review.opendev.org/c/openstack/skyline-apiserver/+/939686\nChange-Id: I18b88e6f8ab5f5e15c4bb0dc4292cb82f6a8107e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57544775fa1c1e99f9dd157d9175f010ddb15c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57544775fa1c1e99f9dd157d9175f010ddb15c0c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cbe81476ef2d7fe164369c515b90e257bac7761a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-01-24 14:49:51.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/939685/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/939685/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/939685/3"}}},"commit":{"parents":[{"commit":"6aaa7d6bb778f142ef7301de9ca51f7cf323bd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6aaa7d6bb778f142ef7301de9ca51f7cf323bd91"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-21 07:53:43.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-01-24 14:49:51.000000000","tz":0},"subject":"Fix Makefile for newer Ubuntu","message":"Fix Makefile for newer Ubuntu\n\n- Pinning setuptools seems no longer to be needed, also breaks due to\n  trying to do a global install\n- Use python3 instead of python binary\n\nDepends-On: https://review.opendev.org/c/openstack/skyline-apiserver/+/935604\nChange-Id: I18b88e6f8ab5f5e15c4bb0dc4292cb82f6a8107e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cbe81476ef2d7fe164369c515b90e257bac7761a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cbe81476ef2d7fe164369c515b90e257bac7761a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
