)]}'
{"id":"openstack%2Fskyline-console~941229","triplet_id":"openstack%2Fskyline-console~master~Icd4d7bf68dc2d11a60ac0378d848d3db37716046","project":"openstack/skyline-console","branch":"master","topic":"rhel-install","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-02-11 12:32:54.000000000","reason":"removed on reply"},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-02-11 17:06:49.000000000","reason":"\u003cGERRIT_ACCOUNT_6282\u003e replied on the change","reason_account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-02-11 17:11:42.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Icd4d7bf68dc2d11a60ac0378d848d3db37716046","subject":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes","status":"MERGED","created":"2025-02-11 10:43:56.000000000","updated":"2025-02-11 17:40:54.000000000","submitted":"2025-02-11 17:39:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941229-rhel-install","meta_rev_id":"d1f7e52e9d4e71976a8720044d317da7b2e210ca","_number":941229,"virtual_id_number":941229,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 17:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-02-11 12:32:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-02-11 17:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-02-11 17:06:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 10:49:03.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-02-11 11:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-11 11:36:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"b398cb546f6fcb0f0cf0fca5f6cc53073e950c8e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-11 10:43:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28c3da772fbb24c39a6e6472028812c15c23d8b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-11 10:45:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3b0deda609655562eeed2d2887e158e34549061d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-11 10:45:58.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"fd7fd2e30bb492779845099e91fd89518c20a0ba","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-02-11 10:49:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"54473aca51e8fa7322888280b83eaf5fc94b0221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 11:34:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/625956d2b9d84871b3d42c4d9eddf57c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ae12579e884a2a84169377ee1e2db0 : SUCCESS in 4m 33s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e7f9a8c3fde6430dbbed6f039a589c23 : SUCCESS in 10m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9ca251850a764d1cae5f42563464a275 : SUCCESS in 4m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5263f5839c5d4e01836001e07404cdbb : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8f1290dd51f44596cfc82d11099ce70a0f56b4e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-11 12:21:59.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a1bcb8f2828096696726debf324a78e812ee9747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-11 12:22:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0e94ad33cf98b837241f539c7f30e23de3e987ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-11 12:22:23.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3114b84d90b2f1e3e02ce67626c07afe6a8c6012","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-02-11 12:32:54.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a2b10a29cfe8113576d468d59dee12422c6671e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:44:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d3298408c144cc8e98fc20a2f768db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4e65bd62ed481fb5922c9a834f18b0 : SUCCESS in 17m 00s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5a1d4f05fb374d09a1684cf4b162f30d : SUCCESS in 11m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/263e58f24fe64b869e7416e35dfdc292 : SUCCESS in 8m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/69d72abc3d06468aa7a571fa9c818e1a : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":5},{"id":"5bc9fefebc8d44be964f13969c415af267654f39","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-11 17:06:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"39eaa831e6ae6310c34e747bdcd80e76adb99517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:07:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"23fcb36e2fbb4433e99d49f43acce67223381997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:39:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/648777191c6047b98158c3f260fa9ea6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3f0b95b0a845eeae3ce27462828e23 : SUCCESS in 4m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/daf80cfae8b44bb795e15cc770809d60 : SUCCESS in 7m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6e378ec351fc43f2837c504fd071b634 : SUCCESS in 6m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8f2902559837425a9003bde1be4d1fcb : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":5},{"id":"f5c4386b821702994cff8ef815681273aace259c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:39:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d1f7e52e9d4e71976a8720044d317da7b2e210ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:40:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dfb006a3c614831b73584eff7cc1fb2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c2b1c87c21c4154ae1af311a54100ae : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f3dc3cba35d773d165b81d8a4b56575b4456ce0a","revisions":{"5d8686099c25ba1bd3e574ab866de4b87da74c80":{"kind":"REWORK","_number":1,"created":"2025-02-11 10:43:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/941229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/941229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/1"}}},"commit":{"parents":[{"commit":"2f59370762ddcfd28ae5f21830cf14ad3f5ac900","subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2f59370762ddcfd28ae5f21830cf14ad3f5ac900"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 10:43:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 10:43:56.000000000","tz":0},"subject":"Docker Install RHEL and CentOS - fix mistakes","message":"Docker Install RHEL and CentOS - fix mistakes\n\nChange-Id: Icd4d7bf68dc2d11a60ac0378d848d3db37716046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d8686099c25ba1bd3e574ab866de4b87da74c80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d8686099c25ba1bd3e574ab866de4b87da74c80"}]},"branch":"refs/heads/master"},"d51d7410190a65817648e5df3ad32a302fb7a376":{"kind":"REWORK","_number":2,"created":"2025-02-11 10:45:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/941229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/941229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/2"}}},"commit":{"parents":[{"commit":"2f59370762ddcfd28ae5f21830cf14ad3f5ac900","subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2f59370762ddcfd28ae5f21830cf14ad3f5ac900"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 10:43:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 10:45:53.000000000","tz":0},"subject":"Docker Install RHEL and CentOS - fix mistakes","message":"Docker Install RHEL and CentOS - fix mistakes\n\nChange-Id: Icd4d7bf68dc2d11a60ac0378d848d3db37716046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d51d7410190a65817648e5df3ad32a302fb7a376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d51d7410190a65817648e5df3ad32a302fb7a376"}]},"branch":"refs/heads/master"},"3fdf5fe21bb5b049b0257613b45d45d712bd5404":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-11 12:21:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/941229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/941229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/3"}}},"commit":{"parents":[{"commit":"2f59370762ddcfd28ae5f21830cf14ad3f5ac900","subject":"Skyline Installation Guide - for RHEL/CentOS - Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2f59370762ddcfd28ae5f21830cf14ad3f5ac900"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 10:43:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 12:21:36.000000000","tz":0},"subject":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes","message":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes\n\nChange-Id: Icd4d7bf68dc2d11a60ac0378d848d3db37716046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3fdf5fe21bb5b049b0257613b45d45d712bd5404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3fdf5fe21bb5b049b0257613b45d45d712bd5404"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dcdf0c76aa25f660faff320f447682eeebacb339":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-11 12:22:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/941229/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/941229/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/4"}}},"commit":{"parents":[{"commit":"b524bc18adac95306b93e3c64cd84a996a18bb92","subject":"Skyline Installation Guide - for RHEL/CentOS - Source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b524bc18adac95306b93e3c64cd84a996a18bb92"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 10:43:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 12:22:04.000000000","tz":0},"subject":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes","message":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes\n\nChange-Id: Icd4d7bf68dc2d11a60ac0378d848d3db37716046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dcdf0c76aa25f660faff320f447682eeebacb339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dcdf0c76aa25f660faff320f447682eeebacb339"}]},"branch":"refs/heads/master","description":"Rebase"},"f3dc3cba35d773d165b81d8a4b56575b4456ce0a":{"kind":"REWORK","_number":5,"created":"2025-02-11 12:22:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/941229/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/941229/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/941229/5"}}},"commit":{"parents":[{"commit":"b524bc18adac95306b93e3c64cd84a996a18bb92","subject":"Skyline Installation Guide - for RHEL/CentOS - Source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b524bc18adac95306b93e3c64cd84a996a18bb92"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 10:43:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-11 12:22:23.000000000","tz":0},"subject":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes","message":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes\n\nChange-Id: Icd4d7bf68dc2d11a60ac0378d848d3db37716046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3dc3cba35d773d165b81d8a4b56575b4456ce0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3dc3cba35d773d165b81d8a4b56575b4456ce0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
