)]}'
{"id":"openstack%2Fskyline-console~941303","triplet_id":"openstack%2Fskyline-console~master~Ia62659e3b2d04dd83df9189d0051f35f3cd16e72","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-02-12 08:21:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia62659e3b2d04dd83df9189d0051f35f3cd16e72","subject":"fix: test failure with new ubuntu \u0026 os 2024.02","status":"MERGED","created":"2025-02-11 23:54:14.000000000","updated":"2025-02-12 08:41:58.000000000","submitted":"2025-02-12 08:21:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941303","meta_rev_id":"3c277bc12065fd7c0c1f2bb6e1f249c9bad8670e","_number":941303,"virtual_id_number":941303,"owner":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 08:21:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":2,"date":"2025-02-12 07:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":1,"date":"2025-02-12 07:07:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 01:20:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 08:17:26.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"}],"messages":[{"id":"e632fe26cbeb877e54482b96108b542a423f48a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-11 23:54:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bd989b0c9852f65656f692ab8196f5fb49d6eee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 01:20:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e41559b4c464eb692a525b8b8e2afa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/222db2eb33224ac4b57a75956d4163e0 : SUCCESS in 4m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c3bc3e8b2f1e4b9f87a3d76e0a9ba053 : SUCCESS in 7m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/376f2ce35292431f9966e49b7e51d6f3 : SUCCESS in 4m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/19b469e11fd7423d9f83269989ea9b22 : SUCCESS in 7m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/862cd2dd59004979a1fdd84ef5584c3e : FAILURE in 23m 29s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bc9d8e6cd45b44339cc6a39ae4e76553 : SUCCESS in 1h 06m 58s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3cacfaf2936941aab6e11df97a6bac5f : SUCCESS in 41m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/25bf6fcf2b41464a8b46edbf5191a4ee : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4a078472d409be2408a5fe03f6757719d86e6f34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-12 01:22:51.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3e0171922eaf05934a9803929fd3a7c3c6d92b56","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-12 02:35:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73f7b7802a8f70664e838aa38d3350e6bcdef874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 02:55:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c9f901eb6564211a80281b65cf7c7e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b479572dfbd14f85a85e8c0e05100e48 : SUCCESS in 4m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a920c565573d431cbe08cddb7f02499f : SUCCESS in 11m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/41402197fc954199b077b0f1a61df456 : SUCCESS in 5m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e09805d1943b41269c450a72769ffb56 : SUCCESS in 9m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8dfdd6ee8e404b7699e269c5020407cc : SUCCESS in 1h 24m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/38e8d8708ace41f8a16b25a46094563e : SUCCESS in 49m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e1c57b586cfa407f9aa1de1c7aac4186 : FAILURE in 59m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3d3c5410bcba49049225ef58ca843e98 : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":2},{"id":"65e9acf162a6244808913f9d7b664ed4792405f2","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-12 02:58:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73c7f92728f346afed56a12bc8c4e2d8c6a9cbc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 04:02:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1c3e6f5cace41558bbc46bbceff70b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2d2f27447846a0a7fa176c8e3f6c4e : SUCCESS in 4m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/aa875cf82ee54c1bab33d1517d5b560f : SUCCESS in 7m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/97a7fa26a1594c1fa8feb1dfb7e6f64a : SUCCESS in 4m 57s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/93c7c5dff0ea4cc08beca3edc40ede5f : SUCCESS in 3m 58s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cab272c36c7e4a49b0bb5b734c03d2dd : SUCCESS in 1h 03m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e4d00b2983ae49ba821a3f6dfa69a615 : SUCCESS in 47m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/573685316bf14abea4dcef3692a7a704 : FAILURE in 51m 07s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/21339da4b3ea4bcaa876f79595dcb8d7 : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9411764c53aa627e126c75911e779b88decd05bf","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-12 04:07:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"452700e29627ef876b513756030695ca3625cf0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 05:27:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75390a12bc6b403a9b854ab004a0d396\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63be360d9244bc28e827b9a4fb25165 : SUCCESS in 3m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d6fdb5efbe2845e8b6b90c9a2e12e2eb : SUCCESS in 8m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1f3051080b8749479dff3244fdbcd44a : SUCCESS in 6m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3d0eba79064b4a27a9c8971d9a2de2d3 : SUCCESS in 6m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8b4ceff8b54a4f1fae07a48c1a156d87 : FAILURE in 1h 17m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/573673177a27458ab1234752565ca3d8 : SUCCESS in 45m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0842c25d3d884db6a40bcbd0bdb1b6d6 : SUCCESS in 40m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/02ef2e0883f54d67b3b049e0217a4c90 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":2},{"id":"cdccbf743f6829e72bafaee52a8f9be4d13bbda7","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-12 05:32:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"506dadc8485d4c33efb4e6ec701076f5f1575781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 07:06:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48ef5acca8714899a402359e93f13eff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b45eac630844f25b18535d58618898b : SUCCESS in 4m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2d3c6ac6875a4362a9221c44a4669fc1 : SUCCESS in 7m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5289e4b9718c47bc9b3abcb0d073bffd : SUCCESS in 4m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2ea30162227e460594374b2db8155b2d : SUCCESS in 4m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3f13e715abd24e83beb66f7ed1bd6b17 : SUCCESS in 1h 07m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e5a4cae2b136425083acea2923cd20b2 : SUCCESS in 1h 03m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2f37232e39514980a3f30d59b151c480 : SUCCESS in 42m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/634749f8a8694c6cb23ec947ffc0ff19 : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":2},{"id":"1a468eb0f1d2f0295f8aad3faa1c08032d57a65b","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-12 07:07:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"278a457dbd572d65ba65ea5a8fd640b3749474c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 07:07:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5eaff5f2b8629578d56164821f690ca065c64ecc","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-02-12 08:17:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5692a84724148424b1addce86aec5bcd7ca243b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 08:21:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c7dddd77e146a7b1d016af74cb06f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7eae3929254586bab52e938db3c752 : SUCCESS in 4m 55s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bca2ab3bf51e4e8da29fec1e330bd8cf : SUCCESS in 4m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d1c0b52d6e5344f1928845e9425b6552 : SUCCESS in 6m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/21dc75f1212b4adb85a5277458e92a66 : SUCCESS in 4m 52s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5c3511c6c3de485a8c5894fc06dc9d93 : SUCCESS in 1h 12m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/543f8a92d11e41859eaa6fb792a0ceac : SUCCESS in 1h 07m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6b1c32c998904cf095fbcc9ed337f1c5 : SUCCESS in 59m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d64325d457ce48e3b454179935b867db : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c833de2a961922c21111a65c66241bb8918627fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 08:21:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a6c36751e1931502cb9fa97a285e31ec9a5533ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 08:22:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/264a45da3886481da01c5fa20cf1e3b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81fb565081df45f791155d88795abc89 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2},{"id":"3c277bc12065fd7c0c1f2bb6e1f249c9bad8670e","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-12 08:41:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47aa72a88bc6fe9682b8f06f67f0280583b9e3d1","revisions":{"a8bd75c177b66ffba08f58570d9dd5553d17ed95":{"kind":"REWORK","_number":1,"created":"2025-02-11 23:54:14.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/03/941303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/941303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/941303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/941303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/941303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/941303/1"}}},"commit":{"parents":[{"commit":"f3dc3cba35d773d165b81d8a4b56575b4456ce0a","subject":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3dc3cba35d773d165b81d8a4b56575b4456ce0a"}]}],"author":{"name":"Wenxiang Wu","email":"wu.wenxiang@99cloud.net","date":"2025-02-11 23:50:14.000000000","tz":480},"committer":{"name":"Wenxiang Wu","email":"wu.wenxiang@99cloud.net","date":"2025-02-11 23:54:11.000000000","tz":480},"subject":"fix: test failure with ubuntu 24.04 \u0026 os 2024.02","message":"fix: test failure with ubuntu 24.04 \u0026 os 2024.02\n\n- Pinning setuptools seems no longer to be needed, also breaks due to\n  trying to do a global install\n- Use python3 instead of python binary\n- Remove all_projects param in list role API\n\nChange-Id: Ia62659e3b2d04dd83df9189d0051f35f3cd16e72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8bd75c177b66ffba08f58570d9dd5553d17ed95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8bd75c177b66ffba08f58570d9dd5553d17ed95"}]},"branch":"refs/heads/master"},"47aa72a88bc6fe9682b8f06f67f0280583b9e3d1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-12 01:22:51.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/03/941303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/941303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/941303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/941303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/941303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/941303/2"}}},"commit":{"parents":[{"commit":"f3dc3cba35d773d165b81d8a4b56575b4456ce0a","subject":"Source Install RHEL or CentOS / Docker Install RHEL and CentOS - fix mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3dc3cba35d773d165b81d8a4b56575b4456ce0a"}]}],"author":{"name":"Wenxiang Wu","email":"wu.wenxiang@99cloud.net","date":"2025-02-11 23:50:14.000000000","tz":480},"committer":{"name":"Wenxiang Wu","email":"wu.wenxiang@99cloud.net","date":"2025-02-12 01:22:25.000000000","tz":480},"subject":"fix: test failure with new ubuntu \u0026 os 2024.02","message":"fix: test failure with new ubuntu \u0026 os 2024.02\n\n- Pinning setuptools seems no longer to be needed, also breaks due to\n  trying to do a global install\n- Use python3 instead of python binary\n- Remove all_projects param in list role API\n\nChange-Id: Ia62659e3b2d04dd83df9189d0051f35f3cd16e72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/47aa72a88bc6fe9682b8f06f67f0280583b9e3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/47aa72a88bc6fe9682b8f06f67f0280583b9e3d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
