)]}'
{"id":"openstack%2Fskyline-console~941801","triplet_id":"openstack%2Fskyline-console~master~Ide83bb7f891d02d18738e44d19ffab64483a2361","project":"openstack/skyline-console","branch":"master","hashtags":[],"change_id":"Ide83bb7f891d02d18738e44d19ffab64483a2361","subject":"fix: build skyline-console-master.tar.gz failed","status":"MERGED","created":"2025-02-15 13:32:37.000000000","updated":"2025-02-16 02:44:41.000000000","submitted":"2025-02-16 02:42:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941801","meta_rev_id":"7f61903f67556b4094fcea08a7708660fb7436d2","_number":941801,"virtual_id_number":941801,"owner":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-16 02:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":2,"date":"2025-02-16 01:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":1,"date":"2025-02-16 01:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-15 14:35:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a52f53a2f54ea787544abaf12b13bc7e6921eb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-15 13:32:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"554c29daf063660a6072f47e167d51b20ae06ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 14:35:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fbb65f45fc9414cae63aba659222180\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578a4af546b64dd480eec50470cb0977 : SUCCESS in 3m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/49cf8975b62e45f8a536112eb4cb9c7c : SUCCESS in 5m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7853f30dcbc34ce9b9a7d2f26c9615d3 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6eade3ca56ff4e1fbe120fff003423fe : SUCCESS in 4m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/dd71fd866f75491d969ecf121e61e5b6 : SUCCESS in 1h 01m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/38d9a4283a584896aab73501d5ec41c7 : SUCCESS in 43m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8111d0aef7854d9faf8cc28e2d3bc16c : SUCCESS in 1h 02m 23s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6384d5a0b2cf4ae48c9a8a372362ded1 : SUCCESS in 35m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a6fffee6eee398f6ad97a0a1d71d2522c9ab5ff8","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-02-16 01:37:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a47dde97d141b49f2831d6ef8921cbff39b3e5a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 01:38:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61a703037831c7243e5a920ba0367b1f8e574b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 02:42:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34b4db4ba62041ffb5a22252070556ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0db22c0de4147cb97742383e71ede9c : SUCCESS in 4m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/02e8ab258ea44777a14757bfa3cfde01 : SUCCESS in 7m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1726a71f90514a088906a52efd3b675b : SUCCESS in 4m 24s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/eb52bb30765946db932d7be5202df05d : SUCCESS in 4m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9f8325c714004291b1aaa501130ea0ce : SUCCESS in 1h 03m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2260a24867d64f83a13c9afc05dcfc1b : SUCCESS in 43m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f8a1bc2504194b6a8efbefc3f6d2051f : SUCCESS in 38m 21s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/43e9e4fc0dd049d983e09e63178c0dc8 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5694a308c33e86b2fed0bff0e65880dcfbea9b2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 02:42:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7f61903f67556b4094fcea08a7708660fb7436d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 02:44:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cb78c7ea5814cacad7b6031d6ad9549\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/637a1fdaf3104fe78d4fe0ccaeefc6c0 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a9936b5468063ddef0d760d128c506df48deefa","revisions":{"8a9936b5468063ddef0d760d128c506df48deefa":{"kind":"REWORK","_number":1,"created":"2025-02-15 13:32:37.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/01/941801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/01/941801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/941801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/941801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/941801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/01/941801/1"}}},"commit":{"parents":[{"commit":"231dddd0656cf772ae85c089d30966c6e7639e1c","subject":"Use make package directly in run.yaml of python-tarball","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/231dddd0656cf772ae85c089d30966c6e7639e1c"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2025-02-15 13:29:48.000000000","tz":480},"committer":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2025-02-15 13:32:32.000000000","tz":480},"subject":"fix: build skyline-console-master.tar.gz failed","message":"fix: build skyline-console-master.tar.gz failed\n\nrefer to: https://zuul.opendev.org/t/openstack/build/1c2e00b50ff440d8a8ad4a2c9b81b93d/console\n\ntarball name like: skyline_console-5.1.0.dev18.tar.gz\n\nCloses-bug: 2098440\nChange-Id: Ide83bb7f891d02d18738e44d19ffab64483a2361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8a9936b5468063ddef0d760d128c506df48deefa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8a9936b5468063ddef0d760d128c506df48deefa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
