)]}'
{"id":"openstack%2Fskyline-console~944339","triplet_id":"openstack%2Fskyline-console~master~Ia51db20fba5b45b1f48f4b86b9034b57d3bf74c4","project":"openstack/skyline-console","branch":"master","topic":"reno-2025.1","attention_set":{},"removed_from_attention_set":{"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-03-21 05:54:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia51db20fba5b45b1f48f4b86b9034b57d3bf74c4","subject":"Update master for stable/2025.1","status":"MERGED","created":"2025-03-14 12:25:03.000000000","updated":"2025-03-21 05:55:51.000000000","submitted":"2025-03-21 05:54:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944339-reno-2025.1","meta_rev_id":"5370ddba6076f411da9b7a582a90e95bf110de3b","_number":944339,"virtual_id_number":944339,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 05:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2025-03-21 05:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-18 13:09:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2025-03-21 05:36:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 12:33:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-18 13:09:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-18 13:09:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2025-03-21 05:36:38.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"68bb6d64fcac1ae75e5166489bc5bf7c7c49be6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-03-14 12:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2efba4fd6005d27a669237f1c77b3f5734464476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 12:33:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713ec1f85d124575943501c85beba561\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/daf7764da20f4df185628d7597e7c17e : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd41e5a08f64e57b65d971457e6dff2 : SUCCESS in 2m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cd3f4587ac584106a0fe177640e39aa8 : SUCCESS in 4m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a27d7362318a4488a0fb429437481b2d : SUCCESS in 4m 54s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/26136e5a69d34979a4505677227e9459 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"82a3e8a0272b264c4f87a2a9be5bb95f68fb67c5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 13:09:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c02f9143f177260beedafddf12cd22ba887fbb6","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-03-21 05:36:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2718302736f3bd42075219cd212f260a0f2c0329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 05:37:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93aee4b9af9db7c4b3fe4e856c417c4cc8ade96a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 05:54:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2746d6ef974cfb8b49502e8cd4d9a9\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbcaebc172cb4d189e59b4a440c813f3 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61361b6ff92948f491b8d74db911fa7c : SUCCESS in 6m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1bc6db29782d42248ba09dedd269d1ab : SUCCESS in 11m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/19ab126e28134adeb846b7cc92d91196 : SUCCESS in 6m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5493202f84d4420a870f386aed51bb30 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"64e73ddac90e51930e12639695a7f4be5388bdbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 05:54:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2746d6ef974cfb8b49502e8cd4d9a9\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbcaebc172cb4d189e59b4a440c813f3 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61361b6ff92948f491b8d74db911fa7c : SUCCESS in 6m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1bc6db29782d42248ba09dedd269d1ab : SUCCESS in 11m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/19ab126e28134adeb846b7cc92d91196 : SUCCESS in 6m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5493202f84d4420a870f386aed51bb30 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"59985d4279e7e43c6a442a97d72b05f1e97ada66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 05:54:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5370ddba6076f411da9b7a582a90e95bf110de3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 05:55:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/560721238109453b802b2f3fdcf2f8b3\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92e49f9ce7ec434dbf324fa96e88c1c5 : SUCCESS in 35s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7be2d5c15342989803183496cecebc : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31dba9f87614b27d8e5adb43ccfafab95c180294","revisions":{"31dba9f87614b27d8e5adb43ccfafab95c180294":{"kind":"REWORK","_number":1,"created":"2025-03-14 12:25:03.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/39/944339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/39/944339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/944339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/944339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/944339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/39/944339/1"}}},"commit":{"parents":[{"commit":"f89e76b36ea840eaac8922920079c0d3b34e4984","subject":"fix: post build tarball failure in ubuntu 20.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f89e76b36ea840eaac8922920079c0d3b34e4984"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 12:25:01.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 12:25:01.000000000","tz":0},"subject":"Update master for stable/2025.1","message":"Update master for stable/2025.1\n\nAdd file to the reno documentation build to show release notes for\nstable/2025.1.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2025.1.\n\nSem-Ver: feature\nChange-Id: Ia51db20fba5b45b1f48f4b86b9034b57d3bf74c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/31dba9f87614b27d8e5adb43ccfafab95c180294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/31dba9f87614b27d8e5adb43ccfafab95c180294"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
