)]}'
{"id":"openstack%2Fskyline-console~945065","triplet_id":"openstack%2Fskyline-console~stable%2F2025.1~Ie56909197a352d933d8a7a8971f515841449031d","project":"openstack/skyline-console","branch":"stable/2025.1","topic":"2025.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-03-21 10:30:16.000000000","reason":"Change was submitted"},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-03-21 10:30:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie56909197a352d933d8a7a8971f515841449031d","subject":"feat: update package for release","status":"MERGED","created":"2025-03-20 07:28:28.000000000","updated":"2025-03-21 10:31:17.000000000","submitted":"2025-03-21 10:30:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945065-2025.1","meta_rev_id":"f00e69a6998b08d74c4ea3cf048628998333a7a7","_number":945065,"virtual_id_number":945065,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 10:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2025-03-21 09:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2025-03-21 09:54:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 07:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-20 09:09:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-03-20 09:09:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2025-03-21 09:54:11.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"297b06b5a043405cc3f45a8dac99dd735c3348c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-03-20 07:28:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54f457d1b94fb83d53b10c5ed5f1744a6095a6bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 07:48:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4b4ea6d9a7049b7b96b104cb2b783d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729dc6c7abf0462db4d298fea2b446cf : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b7f7e3e577f640bc9032bf5e9995645b : SUCCESS in 14m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f94fa6afd6d74a948ef125722ba00946 : SUCCESS in 8m 15s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/aca6565e73ce4c2182d45408eff2afe0 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"04e6c6d9e69a4c2efd2e2050f064b5613eb5f623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"real_author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-03-21 09:53:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_30434\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"_revision_number":2},{"id":"c20b1be18fef98477d57a5b7477569cea30c12b3","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-03-21 09:54:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d545e03d2ade59885ee8010ae53fa5afaf50114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:25:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7738069681314f23bb0ac69f5ffc4630\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4877e4c7d1a4007a625e4d38dcdd4cb : SUCCESS in 7m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/76196c6729c248f1b5702443f5bda427 : SUCCESS in 8m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c159d7e55f0341d6aab43dfbc102a991 : SUCCESS in 7m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c8262745585e427eb604becbfde0b7b4 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7b29eb1d2f60b3d670767c4db6babeb079e1cb45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:25:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04cde3c4db326faca1b03540da5e2013b14e1578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:30:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acb2f5f7e65b4bc8b436a3ea4af74f1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49e8691054f44c6a4e510dd73310abf : SUCCESS in 3m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/342f1bf8847c4280b12b8694de754181 : SUCCESS in 4m 05s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e08e5103b5b44c30a827d34d37f6aff7 : SUCCESS in 3m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/eac71abd01eb49a0ba488ee0827841a5 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2},{"id":"157c4e6bfa791acfc07bfb05a8ba21d47196476f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:30:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f00e69a6998b08d74c4ea3cf048628998333a7a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:31:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f2f6565034e4a72b3fd2fcb075e0d6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd56a6168eb8477fb4868c9b9f46fff2 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"878c56bea931d620ec24de2ddd234ca8d1511c73","revisions":{"3dc8aedc0246bc9d7a4099d4b706a16e33b414ad":{"kind":"REWORK","_number":1,"created":"2025-03-20 07:28:28.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/65/945065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/65/945065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/65/945065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/65/945065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/65/945065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/65/945065/1"}}},"commit":{"parents":[{"commit":"f89e76b36ea840eaac8922920079c0d3b34e4984","subject":"fix: post build tarball failure in ubuntu 20.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f89e76b36ea840eaac8922920079c0d3b34e4984"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-03-20 07:20:18.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-03-20 07:21:21.000000000","tz":0},"subject":"feat: update package for release","message":"feat: update package for release\n\n1. update package for branch 2025.1 release\n2. update gitreview file\u0027s default branch\n\nChange-Id: Ie56909197a352d933d8a7a8971f515841449031d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3dc8aedc0246bc9d7a4099d4b706a16e33b414ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3dc8aedc0246bc9d7a4099d4b706a16e33b414ad"}]},"branch":"refs/heads/stable/2025.1"},"878c56bea931d620ec24de2ddd234ca8d1511c73":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-21 09:53:56.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"real_uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/65/945065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/65/945065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/65/945065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/65/945065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/65/945065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/65/945065/2"}}},"commit":{"parents":[{"commit":"590b105e91401c89bef8df3797e0726e126ba974","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/590b105e91401c89bef8df3797e0726e126ba974"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-03-20 07:20:18.000000000","tz":0},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2025-03-21 09:53:56.000000000","tz":0},"subject":"feat: update package for release","message":"feat: update package for release\n\n1. update package for branch 2025.1 release\n2. update gitreview file\u0027s default branch\n\nChange-Id: Ie56909197a352d933d8a7a8971f515841449031d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/878c56bea931d620ec24de2ddd234ca8d1511c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/878c56bea931d620ec24de2ddd234ca8d1511c73"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
