)]}'
{"id":"openstack%2Fskyline-console~952459","triplet_id":"openstack%2Fskyline-console~master~I14de1e3ad24a6c7c2a19bfa41ccee5fcfdcb83c1","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2025-12-31 19:36:33.000000000","reason":"Change was abandoned"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-12-31 19:36:33.000000000","reason":"Change was abandoned"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2025-12-31 19:36:33.000000000","reason":"Change was abandoned"},"38099":{"account":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"last_update":"2025-12-31 19:36:33.000000000","reason":"Change was abandoned"},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2025-12-31 19:36:33.000000000","reason":"Change was abandoned"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-31 14:42:34.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2025-12-31 19:36:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I14de1e3ad24a6c7c2a19bfa41ccee5fcfdcb83c1","subject":"feat: Hash instance passwords for cloud-init","status":"ABANDONED","created":"2025-06-12 13:27:32.000000000","updated":"2025-12-31 19:36:33.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee1c311a5c5424228a6278b55a490be33309f0f0","_number":952459,"virtual_id_number":952459,"owner":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-31 14:40:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 13:32:00.000000000","updated_by":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2025-06-12 13:32:00.000000000","updated_by":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"},{"updated":"2025-06-12 13:32:00.000000000","updated_by":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2025-06-12 13:32:00.000000000","updated_by":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-06-12 13:32:00.000000000","updated_by":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"reviewer":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"state":"REVIEWER"},{"updated":"2025-06-12 14:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-31 14:42:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"0e86df8b414662787c5de0b88276f7dd07ccd10d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"date":"2025-06-12 13:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"872f7ff7e6558c4b603b26f81f16f065e5fa0b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 14:43:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46e1bb57874a4b68bbb19d4dad4e0560\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0abb31f0f94b9589eec245b8c9e928 : SUCCESS in 6m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/da2d9d2e4fef434b9bb8b5a14cbcd146 : SUCCESS in 4m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e373348775b84b6384c49afd7210c270 : SUCCESS in 5m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/15d8089242cb4776b86b787baf4504d4 : SUCCESS in 5m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c6fe73d6b477434d9a3d45dc59dc2982 : SUCCESS in 1h 10m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6daf54400ef442789752c96f163dcb22 : SUCCESS in 1h 01m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8a565d2f1ea649258087738dcc96c9cf : FAILURE in 58m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/53e10d9d40164aef96b1fb773585f94f : SUCCESS in 27m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ae3c11d7adb6345cee86e9e3d9dcb66107484a23","author":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"date":"2025-06-12 20:42:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0bca2fa992b163f6beca3593234be1b853af503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 22:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdcc070ebf294de4893aff694b2e133b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/501e3b5fda314d829d537310baa0b539 : SUCCESS in 5m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/15553e2b2e6645b99e976c68944a757d : SUCCESS in 10m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b55aa91e91954dfa8ba5029b721e3dbc : SUCCESS in 9m 59s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6e5f8c61770f4740b79b682a495726a9 : SUCCESS in 7m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d1ae5508e3bf41fcab159c20d270883d : SUCCESS in 1h 20m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8a5f2abc82ba431eb464c410e85537ba : SUCCESS in 58m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cec61edaeaeb46e4949181b171119c8a : SUCCESS in 53m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1f304bbaa0e1430daf66d0b48dfe4345 : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":1},{"id":"21bbbfd6eede6ba8c7348488ebd81574df524ca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"date":"2025-07-02 07:21:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"97afdc7636dfc9c94dd29d721ab4c6a3d1567487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 08:51:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4360fc819fd04841840a2e6a076dff47\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62acf896842e415fbd38db98f653d244 : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b2af48cdb8408d8a1c900378239c86 : SUCCESS in 5m 49s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d7890c38177d469ca1a311859b6a66fd : SUCCESS in 9m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ea6a516f479b45208d532d3c8217e231 : SUCCESS in 6m 24s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/98df3807b0a84c12846c2680a39cfe3d : SUCCESS in 6m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5af02ac318f44081bcd249b0b85f22f3 : SUCCESS in 1h 16m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/45812ee8174f4eb7a67cc878fe90dd60 : SUCCESS in 54m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/40c3ab500bd6420e85fa5e1700d8cba0 : SUCCESS in 49m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/cb94d41ae180421c9e98893d9e3bde96 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2ea934141e2065802a9fb51690d2c450abeb884c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 02:26:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"80bae862a5f7f69aa2127e5764820264e02ecf60","author":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"date":"2025-10-01 14:19:10.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"750f395cffa61e7a14fce4d3285a712ae4cf8985","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"date":"2025-10-01 14:20:03.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_38099\u003e\n","accounts_in_message":[{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"}],"_revision_number":2},{"id":"e81a95dc2930eb03e225d00656259cb8891f6b0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 14:40:05.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ac81ec42601b4801b636a33f771228da11f86cc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 14:40:10.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/skyline-console for 952459,3","accounts_in_message":[],"_revision_number":3},{"id":"275ff9791c914cc469d337efc889d5805f7e1b92","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 14:42:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee1c311a5c5424228a6278b55a490be33309f0f0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"date":"2025-12-31 19:36:33.000000000","message":"Abandoned\n\nThis change is no longer needed as the same fix has already been merged elsewhere.\n\nhttps://review.opendev.org/c/openstack/skyline-console/+/962656","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87685758160e03174b5760058dbc27e9d08c4812","revisions":{"5f8c9f85c11b9a63ef91e54e7fe95e6f59c35214":{"kind":"REWORK","_number":1,"created":"2025-06-12 13:27:32.000000000","uploader":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"ref":"refs/changes/59/952459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/952459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/1"}}},"commit":{"parents":[{"commit":"60743e88536d2521f88d38c8cff8a2fa70054e7e","subject":"Merge \"Monkey patch createHash function as workaround for hardcoded md4 hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/60743e88536d2521f88d38c8cff8a2fa70054e7e"}]}],"author":{"name":"huseyin.altun","email":"huseyin.altun@tubitak.gov.tr","date":"2025-06-12 13:15:01.000000000","tz":0},"committer":{"name":"huseyinaltun","email":"huseyin.altun@tubitak.gov.tr","date":"2025-06-12 13:27:13.000000000","tz":0},"subject":"feat: Hash instance passwords for cloud-init","message":"feat: Hash instance passwords for cloud-init\n\nChange-Id: I14de1e3ad24a6c7c2a19bfa41ccee5fcfdcb83c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f8c9f85c11b9a63ef91e54e7fe95e6f59c35214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f8c9f85c11b9a63ef91e54e7fe95e6f59c35214"}]},"branch":"refs/heads/master"},"d14aa287597347752db38a0d9625f54be47e79ea":{"kind":"REWORK","_number":2,"created":"2025-07-02 07:21:44.000000000","uploader":{"_account_id":38099,"name":"Hüseyin Altun","display_name":"Hüseyin Altun","email":"huseyin.altun@tubitak.gov.tr","username":"huseyinaltun"},"ref":"refs/changes/59/952459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/952459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/2"}}},"commit":{"parents":[{"commit":"60743e88536d2521f88d38c8cff8a2fa70054e7e","subject":"Merge \"Monkey patch createHash function as workaround for hardcoded md4 hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/60743e88536d2521f88d38c8cff8a2fa70054e7e"}]}],"author":{"name":"huseyin.altun","email":"huseyin.altun@tubitak.gov.tr","date":"2025-06-12 13:15:01.000000000","tz":0},"committer":{"name":"huseyinaltun","email":"huseyin.altun@tubitak.gov.tr","date":"2025-07-02 07:20:53.000000000","tz":0},"subject":"feat: Hash instance passwords for cloud-init","message":"feat: Hash instance passwords for cloud-init\n\nChange-Id: I14de1e3ad24a6c7c2a19bfa41ccee5fcfdcb83c1\nSigned-off-by: huseyinaltun \u003chuseyin.altun@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d14aa287597347752db38a0d9625f54be47e79ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d14aa287597347752db38a0d9625f54be47e79ea"}]},"branch":"refs/heads/master"},"87685758160e03174b5760058dbc27e9d08c4812":{"kind":"REWORK","_number":3,"created":"2025-12-31 14:40:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/59/952459/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/952459/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/952459/3"}}},"commit":{"parents":[{"commit":"60743e88536d2521f88d38c8cff8a2fa70054e7e","subject":"Merge \"Monkey patch createHash function as workaround for hardcoded md4 hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/60743e88536d2521f88d38c8cff8a2fa70054e7e"}]}],"author":{"name":"huseyin.altun","email":"huseyin.altun@tubitak.gov.tr","date":"2025-06-12 13:15:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 14:39:52.000000000","tz":0},"subject":"feat: Hash instance passwords for cloud-init","message":"feat: Hash instance passwords for cloud-init\n\nChange-Id: I14de1e3ad24a6c7c2a19bfa41ccee5fcfdcb83c1\nSigned-off-by: huseyinaltun \u003chuseyin.altun@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/87685758160e03174b5760058dbc27e9d08c4812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/87685758160e03174b5760058dbc27e9d08c4812"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
