)]}'
{"id":"openstack%2Fskyline-console~953390","triplet_id":"openstack%2Fskyline-console~master~I76e6b18b799e92e64e0bacdb93df5dc8f44978b8","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-08-07 15:33:02.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I76e6b18b799e92e64e0bacdb93df5dc8f44978b8","subject":"feat: Update menu","status":"MERGED","created":"2025-06-26 02:23:38.000000000","updated":"2025-08-08 00:46:58.000000000","submitted":"2025-08-07 17:54:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953390","meta_rev_id":"27fde1b984caf0610f77737036e5c8eb896b326a","_number":953390,"virtual_id_number":953390,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 17:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":2,"date":"2025-08-07 09:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-07 15:33:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":1,"date":"2025-08-07 09:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-26 04:42:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-26 13:38:00.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"CC"},{"updated":"2025-08-07 09:19:09.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2025-08-07 15:33:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"98d4059bc717513f087b391016b3c27071a00fb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-06-26 02:23:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e25f3ae2b010f20f64b3819f15533a25484df8b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-06-26 02:29:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d142a09ea14dc07467b2b7e3c81c4af519e7d8b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-06-26 02:31:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"81696332dec4b50387f70bacaf2c799faabb18b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 04:42:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01936cc692614592adc2c4dc6c2797c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161a3c9b51094d048d190a853d7f1455 : SUCCESS in 5m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7fad3339282f4c609f3fea97d2f025d2 : SUCCESS in 9m 29s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3e05e687f00b4dbeb0b6304f5312239b : SUCCESS in 5m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d1d9fce4c02e41659187d8656368db10 : SUCCESS in 8m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ccac54308ea64a27ae09c0ba15f92615 : SUCCESS in 1h 59m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3f64edebb68c4307bba6c6bfd5634344 : SUCCESS in 35m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b8f8b29fb8d74f7b98d9feface276502 : SUCCESS in 52m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8eb4d0b9fdcb4261bfb60bd578423cdf : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a9bb708add1abc77fd3b37f57deaacc50dbc6b87","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-06-26 13:38:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2fde1e7552757b1616e49ecc018d47feaed4f881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 14:41:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5077829a41394e83a39e0e671b9e2e1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7045045fbec40dea4ab69710c166c13 : SUCCESS in 6m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/11e9910e7a324aa18821fca966399e48 : SUCCESS in 5m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0e4daeecc0bf4b11a2bb3e94e83b85d7 : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8008e4042b2443a18d011c2f41510230 : SUCCESS in 7m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/08c1994b049240e48c167af76a889459 : FAILURE in 55m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/73b333a3967043b19196b1fa2cd4234e : SUCCESS in 58m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d701abd160b342fe8c6b5f0e87f4a85b : SUCCESS in 43m 05s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/599fb8d3b5604816869e84b1509ce2ca : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":3},{"id":"92353767582e30b5f444a0ec572b7c7613873d01","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-06-26 23:54:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f84971541be5966b82ed41169068ed65035253e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 00:59:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f26f9e16ed0144008f822af5404cdb84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ac9bb91ace437ea3affc48bb906c82 : SUCCESS in 3m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2273bed4aa274c37857a92a79c2c3650 : SUCCESS in 6m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4083d67ef5db4108a37e2ab93295900f : SUCCESS in 4m 51s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/aeb5b13b00dc4393a3f73fb26cf85eec : SUCCESS in 4m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/18810f8cf4ef4c9a8460d521eb301daf : SUCCESS in 45m 26s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7dda20844d4a4f80a020e03f7348ec0e : SUCCESS in 59m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d0768981810e4486926783b6a2d8d2d4 : SUCCESS in 28m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b4d09ba9dc8a4b8ca893a84116d80178 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":3},{"id":"333b2960deff187d4383d6c58940c46b80f01a38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-08-07 04:41:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d031af0a354b9e7c80e5eafabbd9125c1f754ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 05:44:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/089cd230ed5240e2a48b53c7fd4cdf45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785aecbb4ccf498dac6155b4967bdd0d : SUCCESS in 5m 33s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5b7400f4be7144f1b9de23b3cdfc3c1e : SUCCESS in 7m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3dae7669335740df81ba3ff6071c4e46 : SUCCESS in 3m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ec7d59571de2420ea5b8e57eaeaaa649 : SUCCESS in 4m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b70be987d350462fa76992ac7f9b3232 : SUCCESS in 43m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9f20f886d7914706943b7e6b288aad35 : SUCCESS in 46m 35s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3c5c418366164f9d99080623a48251d8 : FAILURE in 39m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e7540fa5beca47ff80e2c9ee12902e7d : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":4},{"id":"2bbdc0d2c7e38428540f880af8bbc2b883aeff1a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-08-07 06:24:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"416349c5e091dfe7a33c180c6a7b2f859d3c7581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 07:36:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6a86e8bbd7b4f06b2fec29f77847811\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0acb0cefb8446c994750c0c3314211 : SUCCESS in 5m 31s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8cc23f359aec4ae1bb1cbb8ecc2d8981 : SUCCESS in 10m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/50626db33b78468888489d3031357de5 : SUCCESS in 7m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/38bdd51ea64b4e5294aec431ce96b1a9 : SUCCESS in 7m 36s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/07272e82ee8845dcbd4db6e5f698fd5e : SUCCESS in 1h 09m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a85564bb458f4c019650773828129849 : SUCCESS in 59m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9d39399b8a6c470aae1c4a1a87d9fe3a : SUCCESS in 50m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2a4e37e4df77450fb1b8f3a912dd5c85 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":4},{"id":"e521708643ef70a51652a904f1c9ac075d87dac7","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-08-07 09:19:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"928cad05279e5540300a9b98808c2a73d4177980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:19:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7c3894868d24fb222a77f999917cab24eb138d4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 10:11:15.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00f93c2f243d46cc84097ce462cccd4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33dc97b2416489b817d60a468e7d914 : SUCCESS in 5m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/94023bbb352c480b9d63ed73aef8ea81 : SUCCESS in 8m 27s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4b28435fe4c440f5938862a04b5876c5 : SUCCESS in 6m 57s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/63954aadb2a84e8f94f1a7549359f855 : SUCCESS in 3m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8968c03b61584f3598b527f1c49f774a : FAILURE in 50m 43s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b531b0a2647043739611da4f2a45b6df : SUCCESS in 30m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3698ae26c1e946cb979aa08350dcd8e2 : SUCCESS in 29m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d57248ef6f45478cb87c79cf153d60d6 : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":4},{"id":"3d1bc54262fcc5b0817c83b6785c441f149638ad","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-08-07 13:01:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1fea8db4973dbe883997120f5894759bffd80be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 14:10:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebafd6028df4453786ebe85ff645f665\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538184abd59d49668384d7a29aa236be : SUCCESS in 5m 42s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/006ce4a3803647e889eecf02ce52006d : SUCCESS in 10m 05s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fca02c41b02f411aa9472b3cc99e17a5 : SUCCESS in 3m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7e39c3712d9f4ee290441fbc72097af8 : SUCCESS in 6m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ae08b47cd80c4a2eb2aeeaec9205cb3f : SUCCESS in 44m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5bc54ff2e8a84e349e9f2bbc51a51ff8 : SUCCESS in 1h 03m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6fcaa644aec042aab783609a182dac67 : SUCCESS in 53m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7206ac8da6664a64ac4a7e4393fe0825 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":4},{"id":"17433aa9d7167320ab92e1968fbb2df78c31e525","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 14:10:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e6d3b8a3bb87c7dc0d443d26b0d9b96458f42986","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 15:07:04.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4d7dc3741b94bd7ab71e423ed89a7e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4d507a722b447eb9e3e9f1c7b9d7306 : SUCCESS in 3m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a7782d886a85483184ff7beec06234a2 : SUCCESS in 4m 02s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6d8920fd4bf64662869b366db348d581 : SUCCESS in 3m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9df2786768fe4d72ae5b13da6f5feda5 : SUCCESS in 3m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/af05c6e568dd4b9182304ec5774753d1 : FAILURE in 49m 29s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/934984402c204be2ad00f8c2e0982612 : SUCCESS in 54m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/09f7e116a9d14df498b93bf0a41071dd : FAILURE in 41m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9929d923530145af82577090f31021a9 : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":4},{"id":"3af11349e52a13dc8aa9d75f0ded030bc656609e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-07 15:33:00.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c9ba6f5a64d7d24f78615be7ee18eba76e66c22c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 16:44:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3b3bc383a88444c8461825bd0ca4afe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e04b1fbacb44d1388689a72735b0471 : SUCCESS in 5m 37s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2e00415d9a2f416b89ae65fcdb118884 : SUCCESS in 4m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7a3a1015145a4d888e5a77f782773baa : SUCCESS in 12m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c0cd9666428a4e46b516e8d1833a34be : SUCCESS in 6m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/620cda71d9da495eb77b7d859c699911 : SUCCESS in 1h 08m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1708ea36b59241878f914715363a54a1 : SUCCESS in 58m 57s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c7980f6c9ba248d8944aa46545c47034 : SUCCESS in 50m 41s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a92d3b53bd14449296d658d6672501bd : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":4},{"id":"500483251059b49460468cf971d7423bf4ee1c8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 16:45:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"41e7d28056fd03b21109162ea155c59eb45e1edb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 17:54:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8534403d2ac949de8aafff2b9ec289c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0533751ea071473aa2a162bd22f2b007 : SUCCESS in 6m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/74a4112585ed4d57907bd2cb82e284f2 : SUCCESS in 13m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e2bf2c4ed80244f6aef62e43b30f6c3b : SUCCESS in 11m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ac6ebfced9404b629658ad3d7213a78a : SUCCESS in 7m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ff83af1877854846836afbd0f6e1d634 : SUCCESS in 1h 08m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/29e8416eb5a24ce885ddaaa87eacd351 : SUCCESS in 58m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/85e9dd02d3394dabacd799629853888c : SUCCESS in 50m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ce1371102a0e4a6c9c29b847f03a1863 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":4},{"id":"45c34794490bcdca99b18f4804374e2b1ae8c413","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 17:54:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d0faa7930224c78ce29ffbd76cf1444e1de7a566","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 17:56:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a63a9ea0da74e15bebad1a38899b6bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8f5c74fee74d128d3e53f89177a7e0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4},{"id":"ab9963b5e4b2f9372e29852408be71304f225f82","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-08-08 00:37:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"27fde1b984caf0610f77737036e5c8eb896b326a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-08 00:46:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"30fc0b9e11bfd630a1c3477f64c2308f1301e2bd","revisions":{"d2cb022d168def11f8e79c8f16dac002b27bd365":{"kind":"REWORK","_number":1,"created":"2025-06-26 02:23:38.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/90/953390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/953390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/1"}}},"commit":{"parents":[{"commit":"4a7d5f0315094d7dfb25a1fd892e8e7e875138f7","subject":"Fix errors in portRangeRegex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4a7d5f0315094d7dfb25a1fd892e8e7e875138f7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-06-26 02:22:15.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-06-26 02:22:15.000000000","tz":0},"subject":"feat: Update menu","message":"feat: Update menu\n\nUse the local storage to set the default collapsed status of the menu, so that  user no need to set the menu collapsed every time.\n\nChange-Id: I76e6b18b799e92e64e0bacdb93df5dc8f44978b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d2cb022d168def11f8e79c8f16dac002b27bd365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d2cb022d168def11f8e79c8f16dac002b27bd365"}]},"branch":"refs/heads/master"},"b3b1b2c14f66af5d1157f5c3e02a97206a92b4fa":{"kind":"REWORK","_number":2,"created":"2025-06-26 02:29:19.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/90/953390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/953390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/2"}}},"commit":{"parents":[{"commit":"4a7d5f0315094d7dfb25a1fd892e8e7e875138f7","subject":"Fix errors in portRangeRegex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4a7d5f0315094d7dfb25a1fd892e8e7e875138f7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-06-26 02:22:15.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-06-26 02:29:02.000000000","tz":0},"subject":"feat: Update menu","message":"feat: Update menu\n\nUse the local storage to set the default collapsed status of the menu, so that  user no need to set the menu collapsed every time.\n\nChange-Id: I76e6b18b799e92e64e0bacdb93df5dc8f44978b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b3b1b2c14f66af5d1157f5c3e02a97206a92b4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b3b1b2c14f66af5d1157f5c3e02a97206a92b4fa"}]},"branch":"refs/heads/master"},"59a04dca8172bd7656ac6cb4c92021c97b3cb8d6":{"kind":"REWORK","_number":3,"created":"2025-06-26 02:31:20.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/90/953390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/953390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/3"}}},"commit":{"parents":[{"commit":"4a7d5f0315094d7dfb25a1fd892e8e7e875138f7","subject":"Fix errors in portRangeRegex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4a7d5f0315094d7dfb25a1fd892e8e7e875138f7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-06-26 02:22:15.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-06-26 02:30:59.000000000","tz":0},"subject":"feat: Update menu","message":"feat: Update menu\n\nUse the local storage to set the default collapsed status of the menu, so that  user no need to set the menu collapsed every time.\n\nChange-Id: I76e6b18b799e92e64e0bacdb93df5dc8f44978b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/59a04dca8172bd7656ac6cb4c92021c97b3cb8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/59a04dca8172bd7656ac6cb4c92021c97b3cb8d6"}]},"branch":"refs/heads/master"},"30fc0b9e11bfd630a1c3477f64c2308f1301e2bd":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2025-08-07 04:41:48.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/90/953390/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/953390/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/953390/4"}}},"commit":{"parents":[{"commit":"5124e1706e5595ad2dd57231a833e71d2aea8ac0","subject":"fix: Add X-Openstack-Request-Id for e2e request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5124e1706e5595ad2dd57231a833e71d2aea8ac0"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-06-26 02:22:15.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-08-07 04:41:14.000000000","tz":0},"subject":"feat: Update menu","message":"feat: Update menu\n\nUse the local storage to set the default collapsed status of the menu, so that  user no need to set the menu collapsed every time.\n\nChange-Id: I76e6b18b799e92e64e0bacdb93df5dc8f44978b8\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/30fc0b9e11bfd630a1c3477f64c2308f1301e2bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/30fc0b9e11bfd630a1c3477f64c2308f1301e2bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
