)]}'
{"id":"openstack%2Fskyline-console~956616","triplet_id":"openstack%2Fskyline-console~master~I443f928efeb9157bbfe952740c48d721f71e66b2","project":"openstack/skyline-console","branch":"master","topic":"e2e","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-08-07 04:38:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I443f928efeb9157bbfe952740c48d721f71e66b2","subject":"fix: Add X-Openstack-Request-Id for e2e request","status":"MERGED","created":"2025-08-06 01:47:27.000000000","updated":"2025-08-07 04:39:52.000000000","submitted":"2025-08-07 04:38:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956616-e2e","meta_rev_id":"ee673db6e15cf66e7a56b77723eeaf5762bda2bb","_number":956616,"virtual_id_number":956616,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 04:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":2,"date":"2025-08-07 03:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":1,"date":"2025-08-07 03:05:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 02:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-07 03:05:20.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"1ed6e6c102d6fa887cbb6e70e74f84893b8d4ac5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-08-06 01:47:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c37aa06859c60a128f70bf61cb8041611866cde5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 02:49:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a979ac7b2eb4628bff11b90f44b1c3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b2b6824474447e9683c78dccdbbed7 : SUCCESS in 6m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/47f404c58d4141619323fe6f6be21796 : SUCCESS in 4m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8126395d06bd45d5b193bc15036d0894 : SUCCESS in 3m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5b92dffa7be148f0b414d2b72eaefc86 : SUCCESS in 7m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/04d7b4fced4d4e32b74fb04a81128568 : FAILURE in 50m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d7d1c6e3252e4e5794d74592cee56c5b : FAILURE in 39m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7103e044f01842f8af14375ce3a58ffd : SUCCESS in 50m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6cee1e1b59d5448996235aba806b3be1 : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3ce1c1a1e7537e535c8efec15995b2a220b2c179","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-08-06 06:10:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3594eaafbad953743d9a79e95138c8c2ff922cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 07:46:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ac877057d71449fb732bbcde9485cf6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05666fb2e04b44249a567b39a218967b : SUCCESS in 5m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/51f15aa915444b29ab5dae3838b2aca5 : SUCCESS in 8m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f1563db4775b4b878ad33ec3d7ba7b9a : SUCCESS in 6m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/89e5134d4c5645a0bb2bdcb1c22e68b9 : SUCCESS in 4m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/defa39365fd34763802661f49f03042c : SUCCESS in 1h 21m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d9e4c6596f9e4a4fafc47986991eceb4 : FAILURE in 1h 35m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ee4b670939d5433faa8f97d90d426ffa : SUCCESS in 28m 35s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9f5bb87f32bb428e96df218da1f2e71a : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ed60da0ea730338698ff0948589f366570dd1bbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-08-06 08:55:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"37abd5bcdeadd6f4a0da8b1a57b84a241da5012b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 10:01:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76ba36ca182647f49283ea89df2546f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef02b75155224c90abc854d98dda43b8 : SUCCESS in 7m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/62a43dcd1b344c2090467027d7b8225d : SUCCESS in 10m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e428b7a5a8bc4a5e8115cee0b2b67d2c : SUCCESS in 7m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/292494eab6b249c0a64fb0f9045248e7 : SUCCESS in 5m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/19db22d45f9c415b90006e143a35f8b9 : SUCCESS in 1h 04m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/143e2fb5847f4c1d8f88cc6bb2baccfa : SUCCESS in 1h 01m 22s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/64b26354b53c4ba3ac26aa55be702b83 : SUCCESS in 47m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e84524ebcb1a4d00a3ac6a6c55f7e3a4 : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":2},{"id":"cf31b888cc8d3c9ef2e30b5c0c64b51c987d1a06","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-08-07 03:05:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c66389415955c9c975c2a13c110d4a906b29c09f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 03:05:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"867757fdd5acbdcff2439c09257cc6b1a79fee1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 04:38:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/664431366add458abfe401ef54bdfaae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d5a6ff0a064ed681ffaab18e369413 : SUCCESS in 7m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7f7d700e463c4235ac74cb03812d56a4 : SUCCESS in 10m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ed1135a40a83406ab481ec43ba92a0aa : SUCCESS in 6m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b976914772724b33a35e98a58e86c033 : SUCCESS in 6m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8f9e5d2b3da0404292c418eacd3fefe7 : SUCCESS in 1h 32m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d963223236684c8296bbc75a923b1bd7 : SUCCESS in 57m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/37e5d8fc12924d24998a0b7bf675e2bc : SUCCESS in 52m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d7ed3791c9a2408d87f389fbbd6cc8a8 : SUCCESS in 49m 27s","accounts_in_message":[],"_revision_number":2},{"id":"5835d959b84ec5b5584b78cb0a143f1deda55e15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 04:38:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ee673db6e15cf66e7a56b77723eeaf5762bda2bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 04:39:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52cf724073ca483b96940bbda7852dbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c90c73767bb433ab45ba9912f824247 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5124e1706e5595ad2dd57231a833e71d2aea8ac0","revisions":{"30c6ff83d6813e358f1fff1c4498b2ad72cfeac8":{"kind":"REWORK","_number":1,"created":"2025-08-06 01:47:27.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/16/956616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/956616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/956616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/956616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/956616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/956616/1"}}},"commit":{"parents":[{"commit":"4a7d5f0315094d7dfb25a1fd892e8e7e875138f7","subject":"Fix errors in portRangeRegex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4a7d5f0315094d7dfb25a1fd892e8e7e875138f7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-08-06 01:46:03.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-08-06 01:46:48.000000000","tz":0},"subject":"fix: Add X-Openstack-Request-Id for e2e request","message":"fix: Add X-Openstack-Request-Id for e2e request\n\nAdd X-Openstack-Request-Id for the e2e login request which is a necessary header for requesting the skyline api\n\nChange-Id: I443f928efeb9157bbfe952740c48d721f71e66b2\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/30c6ff83d6813e358f1fff1c4498b2ad72cfeac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/30c6ff83d6813e358f1fff1c4498b2ad72cfeac8"}]},"branch":"refs/heads/master"},"5124e1706e5595ad2dd57231a833e71d2aea8ac0":{"kind":"REWORK","_number":2,"created":"2025-08-06 08:55:01.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/16/956616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/956616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/956616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/956616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/956616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/956616/2"}}},"commit":{"parents":[{"commit":"4a7d5f0315094d7dfb25a1fd892e8e7e875138f7","subject":"Fix errors in portRangeRegex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4a7d5f0315094d7dfb25a1fd892e8e7e875138f7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-08-06 01:46:03.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-08-06 08:54:28.000000000","tz":0},"subject":"fix: Add X-Openstack-Request-Id for e2e request","message":"fix: Add X-Openstack-Request-Id for e2e request\n\n1. Add X-Openstack-Request-Id for the e2e login request which is a necessary header for requesting the skyline api\n2. Update e2e case\n\nChange-Id: I443f928efeb9157bbfe952740c48d721f71e66b2\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5124e1706e5595ad2dd57231a833e71d2aea8ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5124e1706e5595ad2dd57231a833e71d2aea8ac0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
