)]}'
{"id":"openstack%2Fskyline-console~956980","triplet_id":"openstack%2Fskyline-console~master~I30909cb14c9ac932a323cf00a285e5f925f07fc1","project":"openstack/skyline-console","branch":"master","hashtags":[],"change_id":"I30909cb14c9ac932a323cf00a285e5f925f07fc1","subject":"fix: Need trim filter string","status":"MERGED","created":"2025-08-09 04:24:51.000000000","updated":"2025-08-09 08:20:00.000000000","submitted":"2025-08-09 08:18:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956980","meta_rev_id":"b81a60d145887810a997e6cd76da4fe97f4f7231","_number":956980,"virtual_id_number":956980,"owner":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-09 08:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":2,"date":"2025-08-09 06:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":1,"date":"2025-08-09 06:49:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-09 06:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca5305deee0dc5431e9dd5852edc52030f56662c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-08-09 04:24:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8f8b4332af7ff7769b25f9b2a48ab41dc6d3e1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-08-09 05:11:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08da76331e32b27c0597c1abfc17786e43520c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 06:35:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fe1cb56885f4561ace272e668c3aa7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/172c6ce728984fe480bccaf7e6ce1a87 : SUCCESS in 5m 11s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5c0e61aaa9d64fafae6fdddd231fd150 : SUCCESS in 6m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/666bef80bfee402e98f4cb53b8eb6af1 : SUCCESS in 4m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/95a7f79d02594973839df7a76adefaaf : SUCCESS in 4m 58s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c952b2c0dd66459b9c8e5e75fbbaaf87 : SUCCESS in 1h 19m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ac24ed1dcf5142fa991c602b0eb2721f : SUCCESS in 44m 35s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/08151c1abf6245c4b9de356a17d85436 : SUCCESS in 52m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2ff361ff123549bda2f4c25838c89263 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":2},{"id":"cf6b90a7e1912d6aab393978b80ae2c3c3e51b66","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-08-09 06:49:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f19d3b1f09e9abe7fcdb8bb34ecd29ba5168ebcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 06:50:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e25206a72e3575542e8c0916f5cf4765d187710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 08:18:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/787abc1d61bd4a5ba344c67d2b3c9c78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1efceddee323461ebbcc44f1182f772a : SUCCESS in 6m 59s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/83c3e7faf68b400eafa68ae1f3037af4 : SUCCESS in 8m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1bf970b9ec78442baea41bdb3dc30355 : SUCCESS in 6m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2968396eabeb4329b76ffb388f262fc3 : SUCCESS in 6m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/848cc5867b0640a0ac98b96c7f58c73c : SUCCESS in 1h 21m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2baad6d239c44763bab800ee356bba98 : SUCCESS in 57m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/017d06656efd467c93c05780d4dc5f47 : SUCCESS in 52m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bb6e89e9d917477393f6e8cb298a5894 : SUCCESS in 48m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ad7ee340657c4e1e64c7c1a72092012965f28ce8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 08:18:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b81a60d145887810a997e6cd76da4fe97f4f7231","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-09 08:20:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c744f230e8f24603aa41d43109c7bdd6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98af40c1125544449e0f82527bc5aa48 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52d82d2eda29aff90f2e568ccb0ef9506d233bb5","revisions":{"307b64a09eb67e647a698f70ee4163c4da614d9c":{"kind":"REWORK","_number":1,"created":"2025-08-09 04:24:51.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/80/956980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/956980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/956980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/956980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/956980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/956980/1"}}},"commit":{"parents":[{"commit":"164a99413dc9e47a47ba338cfb1267205c66dece","subject":"feat: Update auto refresh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/164a99413dc9e47a47ba338cfb1267205c66dece"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2025-08-09 04:23:22.000000000","tz":480},"committer":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2025-08-09 04:23:22.000000000","tz":480},"subject":"fix: Need trim filter string","message":"fix: Need trim filter string\n\nChange-Id: I30909cb14c9ac932a323cf00a285e5f925f07fc1\nSigned-off-by: Wu Wenxiang \u003cwu.wenxiang@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/307b64a09eb67e647a698f70ee4163c4da614d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/307b64a09eb67e647a698f70ee4163c4da614d9c"}]},"branch":"refs/heads/master"},"52d82d2eda29aff90f2e568ccb0ef9506d233bb5":{"kind":"REWORK","_number":2,"created":"2025-08-09 05:11:41.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/80/956980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/956980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/956980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/956980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/956980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/956980/2"}}},"commit":{"parents":[{"commit":"164a99413dc9e47a47ba338cfb1267205c66dece","subject":"feat: Update auto refresh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/164a99413dc9e47a47ba338cfb1267205c66dece"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2025-08-09 04:23:22.000000000","tz":480},"committer":{"name":"Wu Wenxiang","email":"wu.wenxiang@99cloud.net","date":"2025-08-09 05:11:26.000000000","tz":480},"subject":"fix: Need trim filter string","message":"fix: Need trim filter string\n\nChange-Id: I30909cb14c9ac932a323cf00a285e5f925f07fc1\nSigned-off-by: Wu Wenxiang \u003cwu.wenxiang@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/52d82d2eda29aff90f2e568ccb0ef9506d233bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/52d82d2eda29aff90f2e568ccb0ef9506d233bb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
