)]}'
{"id":"openstack%2Fskyline-console~958168","triplet_id":"openstack%2Fskyline-console~master~I8bf452129ecfb301f8e94037007ef821e3a975fe","project":"openstack/skyline-console","branch":"master","topic":"resize","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-09-28 04:18:56.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-09-28 03:12:17.000000000","reason":"removed on reply"},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2025-09-28 04:18:56.000000000","reason":"Change was submitted"},"7353":{"account":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"last_update":"2025-09-28 04:18:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8bf452129ecfb301f8e94037007ef821e3a975fe","subject":"fix: Notify users to Confirm/Revert after VM resize","status":"MERGED","created":"2025-08-21 10:30:28.000000000","updated":"2025-09-28 04:19:44.000000000","submitted":"2025-09-28 04:18:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958168-resize","meta_rev_id":"45543a39501577828bfc1896780b06e29ae46d9a","_number":958168,"virtual_id_number":958168,"owner":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-28 04:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-09-28 03:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-09-28 03:12:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 12:01:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-25 13:35:07.000000000","updated_by":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2025-08-25 13:35:07.000000000","updated_by":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-09-23 03:20:50.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"464c271404df41f0cf341468ce9425c24fae42bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-21 10:30:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa26535084f1a48ebf5dd56294ac915030520e26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-21 10:31:51.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"198e9123ce401e0130e5335a801c77d228236985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 12:01:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17b0b14e7a084282a34636e7b7059b7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c28ef742df9405ea586c7979f5525b4 : SUCCESS in 4m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2b324a2de924452196e11dfc11c0f911 : SUCCESS in 4m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e5ec0468774f4ff8b7a027c6686224b1 : SUCCESS in 7m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8e2a3cee486d41cfb8309966103a96c6 : SUCCESS in 7m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f5cd2a9eec5e41faa9842f07ac9ad154 : SUCCESS in 1h 22m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8f65adc2c1304f2188238896f265d4e3 : SUCCESS in 57m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/99a5841815464fc8b18403a2036e81e4 : SUCCESS in 33m 41s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ce5e5622b8b44b85857308fc07a1ad19 : FAILURE in 51m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6fb6309df9a397019a26292aafb9f85e481acad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-25 09:39:19.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"101c35538d3c869a9aa8b4894c67ef89ec929417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-25 09:40:52.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4ab59b4f2ab3e0089184b9aa9d1612c6f7bf2399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 11:01:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49cd9dbeacf1455897dda510284369ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353e7b7234124e20bf3cca9558d1b3f2 : SUCCESS in 5m 31s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/32f30d15e505422db028d2ea9443ca85 : SUCCESS in 9m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ba68e943153449a1bf7dea97d5c45467 : SUCCESS in 4m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/be86336e9b724cdaa0d4d71d8193dec1 : SUCCESS in 4m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6d822270deff45e5859027787340c7b6 : SUCCESS in 1h 18m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2113ab9abe384fbeb2592d6f2abeed4c : SUCCESS in 57m 07s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0524b39b79ff48e4a1c142da333c1779 : SUCCESS in 52m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/77cb472d38e84e139c12a47194d5f896 : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":4},{"id":"516b8db4e654ad257ca237e3bf8b4af0fb447be1","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-09-13 23:45:54.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"73c6d720f7976afb7548c2ccb6254aad38e8775e","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-09-13 23:46:30.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5e8a98424ab6c524458ce7592cb6a36dde2ba136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 06:52:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4ef04b5b35cde38eb7280bdf40c941e9489c6a35","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 07:07:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dcc0dc62aa1fd20d100fbe6507b5d49ba8d63761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 07:23:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/053809bd055744709890f48dcc929c36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f699b05b42ab454b98ca95459163b3c5 : SUCCESS in 4m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0de5fef2eaf643b7baa89616d94dc5c4 : SUCCESS in 8m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2e452e64f91b46dd880c910b11d37550 : SUCCESS in 7m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c4796f82ffe74fc8953ff3b9b31df1e0 : SUCCESS in 7m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c06dfa0a4873488f8c1c8994fd678cdb : RETRY_LIMIT in 5m 18s","accounts_in_message":[],"_revision_number":5},{"id":"40cfe2d8b38c2a74016933ab9fe7c7084816edfc","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-23 03:20:50.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"71bb083a1d36234c9988e91eb716e5bdd4daf982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-24 03:27:36.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"9f3425e8bd58485379456d49bacc0d7c749d3d48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 04:04:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46afd595ec164d6a921e6782328c01d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94baf64eb9d941f8ac29e63f18d01bf3 : SUCCESS in 5m 27s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c66e37b0c5784733a66cbd4acd561ebc : SUCCESS in 7m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/550abc748850453b849b98619a95f695 : SUCCESS in 5m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3963513ae9234fd2930b761286b493c6 : SUCCESS in 3m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5e69402767e649b4b7aff2edf8ec1c43 : RETRY_LIMIT in 6m 41s","accounts_in_message":[],"_revision_number":6},{"id":"f7875bdbddd66cd1f711835728f12e38d25cd931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-26 11:31:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a82498b58f216cfc19c2d573a31185d85a4d14bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 12:53:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/027c0ad49c36488293961ddf7bc163fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf9c8172d504461b0ecdebd04bbd4b7 : SUCCESS in 7m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/073cf58b338c4e76a3cc28a2fdb3dfc8 : SUCCESS in 6m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/01dee6c04de54e07908d184908ef4648 : SUCCESS in 7m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/83f514e2dd0f4e3299c0f4c44a2d1a11 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/473ecc96788246829b63fcb1d6e6d9aa : SUCCESS in 1h 19m 33s","accounts_in_message":[],"_revision_number":7},{"id":"26b4024b07212303780cadcc4bdd225a9a5799c5","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 03:12:17.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"aa692105a351440503a47afb807829899811de70","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 03:12:24.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c7310d67930e3d54f1acac5425a7e2033c89bdd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 03:12:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"12153b2182201d4c98669218908251179f894d93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 04:18:55.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97badeb5c9314ea8a8b7d6a17d6d4d1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c496c93bfa4a52b9f9d9e8a170f927 : SUCCESS in 3m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/edc818a9b312482297888ca340015eb5 : SUCCESS in 7m 27s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/909692bc9cf34771b365947c953f5a74 : SUCCESS in 3m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d66d9d53df864da7b90b3f57202a93cc : SUCCESS in 8m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f44a7b525b3a4b88b71164ec18b781d7 : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":7},{"id":"37e4578d5a55cc05aaba250dceb32495f0391250","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 04:18:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"45543a39501577828bfc1896780b06e29ae46d9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 04:19:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91d25e0926648218e0fac5bb69e8193\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e38aafc19754ee1900c5ab4c43716f5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a6f359bf5bcf6e0aa0fd7b0341d5b0330ef744ae","revisions":{"180786393945c2ea8b6e02043fe5cf4c53f91f91":{"kind":"REWORK","_number":1,"created":"2025-08-21 10:30:28.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/68/958168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/958168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 10:19:35.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 10:29:09.000000000","tz":330},"subject":"fix: Notify users to Confirm/Revert after VM resize","message":"fix: Notify users to Confirm/Revert after VM resize\n\nWhen a VM is resized or migrated, the UI only shows \"Resizing or Migrating\"\nwithout indicating that user action is required. The server stays in this\nstate until the user confirms or reverts.\n\nThis change adds a notification prompting the user to confirm or revert.\n\nChange-Id: I8bf452129ecfb301f8e94037007ef821e3a975fe\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/180786393945c2ea8b6e02043fe5cf4c53f91f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/180786393945c2ea8b6e02043fe5cf4c53f91f91"}]},"branch":"refs/heads/master"},"3304f5bfc2b072867533bdfff45b84e189e9ff77":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-21 10:31:51.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/68/958168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/958168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/2"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 10:19:35.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 10:31:51.000000000","tz":0},"subject":"fix: Notify users to Confirm/Revert after VM resize","message":"fix: Notify users to Confirm/Revert after VM resize\n\nWhen a VM is resized or migrated, the UI only shows \"Resizing or Migrating\" without indicating that user action is required. The server stays in this state until the user confirms or reverts.\n\nThis change adds a notification prompting the user to confirm or revert.\n\nChange-Id: I8bf452129ecfb301f8e94037007ef821e3a975fe\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3304f5bfc2b072867533bdfff45b84e189e9ff77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3304f5bfc2b072867533bdfff45b84e189e9ff77"}]},"branch":"refs/heads/master","description":"Edit commit message"},"10f77fb2c053a79702b4007be5e55db795e32f39":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-08-25 09:39:19.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/68/958168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/958168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/3"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 10:19:35.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-25 09:39:02.000000000","tz":330},"subject":"fix: Notify users to Confirm/Revert after VM resize","message":"fix: Notify users to Confirm/Revert after VM resize\n\nWhen a VM is resized or migrated, the UI only shows \"Resizing or Migrating\"\nwithout indicating that user action is required. The server stays in this\nstate until the user confirms or reverts.\n\nThis change adds a notification prompting the user to confirm or revert.\n\nChange-Id: I8bf452129ecfb301f8e94037007ef821e3a975fe\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/10f77fb2c053a79702b4007be5e55db795e32f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/10f77fb2c053a79702b4007be5e55db795e32f39"}]},"branch":"refs/heads/master"},"d95993f6d7a0a8f38d2e8548de4947b3b33da29c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-08-25 09:40:52.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/68/958168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/958168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/4"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 10:19:35.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-25 09:40:52.000000000","tz":0},"subject":"fix: Notify users to Confirm/Revert after VM resize","message":"fix: Notify users to Confirm/Revert after VM resize\n\nWhen a VM is resized or migrated, the UI only shows \"Resizing or Migrating\" without indicating that user action is required. The server stays in this state until the user confirms or reverts.\n\nThis change adds a notification prompting the user to confirm or revert.\n\nChange-Id: I8bf452129ecfb301f8e94037007ef821e3a975fe\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d95993f6d7a0a8f38d2e8548de4947b3b33da29c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d95993f6d7a0a8f38d2e8548de4947b3b33da29c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"af000afeb54c960caeca9d20e697af78639d1e27":{"kind":"REWORK","_number":5,"created":"2025-09-15 06:52:58.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/68/958168/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/958168/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/5"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 10:19:35.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 06:52:41.000000000","tz":330},"subject":"fix: Notify users to Confirm/Revert after VM resize","message":"fix: Notify users to Confirm/Revert after VM resize\n\nWhen a VM is resized or migrated, the UI only shows \"Resizing or Migrating\"\nwithout indicating that user action is required. The server stays in this\nstate until the user confirms or reverts.\n\nThis change adds a notification prompting the user to confirm or revert.\n\nChange-Id: I8bf452129ecfb301f8e94037007ef821e3a975fe\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af000afeb54c960caeca9d20e697af78639d1e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af000afeb54c960caeca9d20e697af78639d1e27"}]},"branch":"refs/heads/master"},"9ab9f32436a7fb6a8a11188b80593e21a3181b5a":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-09-24 03:27:36.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/68/958168/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/958168/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/6"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 10:19:35.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-24 03:27:36.000000000","tz":0},"subject":"fix: Notify users to Confirm/Revert after VM resize","message":"fix: Notify users to Confirm/Revert after VM resize\n\nWhen a VM is resized or migrated, the UI only shows \"Resizing or Migrating\" without indicating that user action is required. The server stays in this state until the user confirms or reverts.\n\nThis change adds a notification prompting the user to confirm or revert.\n\nChange-Id: I8bf452129ecfb301f8e94037007ef821e3a975fe\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9ab9f32436a7fb6a8a11188b80593e21a3181b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9ab9f32436a7fb6a8a11188b80593e21a3181b5a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a6f359bf5bcf6e0aa0fd7b0341d5b0330ef744ae":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-26 11:31:27.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/68/958168/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/958168/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/958168/7"}}},"commit":{"parents":[{"commit":"38fbc9258aaa8f06bbf07c120a336f209c68366f","subject":"fix: invalid vars syntax in playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38fbc9258aaa8f06bbf07c120a336f209c68366f"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-26 11:30:37.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-26 11:30:57.000000000","tz":330},"subject":"fix: Notify users to Confirm/Revert after VM resize","message":"fix: Notify users to Confirm/Revert after VM resize\n\nWhen a VM is resized or migrated, the UI only shows \"Resizing or Migrating\" without indicating that user action is required. The server stays in this state until the user confirms or reverts.\n\nThis change adds a notification prompting the user to confirm or revert.\n\nChange-Id: I8bf452129ecfb301f8e94037007ef821e3a975fe\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a6f359bf5bcf6e0aa0fd7b0341d5b0330ef744ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a6f359bf5bcf6e0aa0fd7b0341d5b0330ef744ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
