)]}'
{"id":"openstack%2Fskyline-console~958194","triplet_id":"openstack%2Fskyline-console~master~I83b8b531363aec94634254b35b4ac3b87bd2e8e6","project":"openstack/skyline-console","branch":"master","topic":"lb-member","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-09-30 04:04:39.000000000","reason":"removed on reply"},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2025-09-30 08:32:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83b8b531363aec94634254b35b4ac3b87bd2e8e6","subject":"feat: Add enable/disable action for load balancer member and display provisioning/operating status","status":"MERGED","created":"2025-08-21 12:55:49.000000000","updated":"2025-09-30 08:33:08.000000000","submitted":"2025-09-30 08:32:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958194-lb-member","meta_rev_id":"e527192120736044a1c9eb4d1406d79a479ebf4b","_number":958194,"virtual_id_number":958194,"owner":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-30 08:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-09-30 04:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-09-30 04:04:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 13:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 03:05:57.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"b2173410e8abd3ae18500ffad9bf4527c5e478cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-21 12:55:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6148b2e222aa3a6d75c61340f6a46e9d6f11ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 13:23:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72429b8be8504cb4883367c8f3e2070a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389474537c4c4ab59786227feeff3faf : SUCCESS in 5m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/81bb798a84ca4121905530d1d6e90886 : SUCCESS in 10m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/16089a55f8324f52b87a1425fc114ed0 : SUCCESS in 7m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fadd6a38850e4c009792e181df08035f : SUCCESS in 6m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d5a555e0f9204c98851f19afac2d0032 : RETRY_LIMIT in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"403c05471be84baf519934a8e79a9b171c34f22d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-26 09:51:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"56419746ece695b76922f08ebd1380aea6089145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 10:25:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee31cb6c01184507894eea85cba8acd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae50173913fe48ee9bf05e842758856d : SUCCESS in 4m 00s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3dd5668410354519a854cda6ae6e12ae : SUCCESS in 6m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ce6029c167644c2da1a7aa42fad32cef : FAILURE in 2m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/159d11a4fa984aa5a85f633cc91b2e21 : SUCCESS in 3m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9674bcbe9e4047dc9e2788c75217a06d : RETRY_LIMIT in 7m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/613c4cd2fe1e4adba60b5b968049c39c : RETRY_LIMIT in 6m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/efc201daf03e48c3b8a052ad57cda0dd : RETRY_LIMIT in 6m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/210924ac9ff54add8c252f04f1c875ea : RETRY_LIMIT in 9m 03s","accounts_in_message":[],"_revision_number":2},{"id":"02be95ac2ddea60e099a1f7925808d6dc8e71ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-26 12:08:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"efbb73a0c8891083485611ce264a979c4c5ae511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 13:18:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea982346ea7247c6a98e8a5a2af2c9b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88a3ac145a34fca9919836c1551ab51 : SUCCESS in 3m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/694b143d72474e1aa8c912e0373152f2 : SUCCESS in 10m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/30d7a48b4ffb494db2920cc89d58a2ff : SUCCESS in 7m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3c68a6e16d5c4379aa75da9bf7ed05f6 : SUCCESS in 5m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6e61d0be0b2b429d837b59c5fbfa04a1 : SUCCESS in 1h 07m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/30d2e23a2a324072b6e75a6bac779224 : SUCCESS in 1h 00m 07s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0c268f185aed45a1b82b91362323b6dc : SUCCESS in 53m 19s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a1f9909efe644e6392c085f3c63c3a56 : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":3},{"id":"bea75f2230b4364372cec68789428ca94f1b0a4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 07:28:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"25171e48932ce355c3c959adb3939166ac975a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 08:12:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0840a15aa6484aa9a0dfe46e1c571d50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f954c3f36a404ebb0888e2258e54ac : SUCCESS in 4m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b6aba0b1fef245baac7b1a49d107287c : SUCCESS in 10m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b97f1e156fb04eeabedb7b162db259b1 : SUCCESS in 7m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/37855028d1094a7c9bc080cdb7e601a2 : SUCCESS in 7m 05s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8973a57691ef40c39ef8c0e59d394919 : RETRY_LIMIT in 9m 27s","accounts_in_message":[],"_revision_number":4},{"id":"c0b4b6f649cda013fe224351d225ad79ecc91190","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-23 03:05:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"884a9ad7f08ca1fc89e238ba8881be8bc7ed9d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-24 10:47:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e9627c7da9e2ce73abe53a213a42be95479b8b12","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-24 10:47:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dca5205206d2fe2432ee49246dd493dc19c45aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 11:07:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68a772d536964acb90b359cb34794c51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d154b189a614fbfbd4f2b01140046e4 : SUCCESS in 3m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b452195d84754721adcc0a9dd6e76d96 : SUCCESS in 4m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ea754515c6e94d808c20499f386e13cb : SUCCESS in 3m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/912dbb281c2b479496e04a97cd390a56 : SUCCESS in 3m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b297e8b026694764954ee71c90f6d889 : RETRY_LIMIT in 3m 59s","accounts_in_message":[],"_revision_number":5},{"id":"e7ce135495461823007d0166f04a0768c550ae19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-24 11:36:40.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5795a6da7727b4338574c6fc4ede257954b93277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 11:54:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89f24a226c94472e8de6dd9fdf88d0cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78ed593de7f45459af91b87faaf465f : SUCCESS in 4m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7b1592fd91af44118bde74198a2fa0ed : SUCCESS in 6m 02s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e691bafadac44e019f7f5c891c11efe7 : SUCCESS in 7m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8ec54306c4e241db90977034acd83241 : SUCCESS in 4m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5b87b3c2888e4d489a42cfefc2be09ca : RETRY_LIMIT in 3m 52s","accounts_in_message":[],"_revision_number":6},{"id":"4cb63b00ddcea370eb957891f7af63c0b9f1fa5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-26 10:16:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"349e0708e75481fcd379fe6d98cc29f1887185cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 11:17:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2c55f573f424a8f9146821ccd1202cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53faafc469b4fbda59048e77c8ad293 : SUCCESS in 3m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6e083feb447e4a87b0993d299c25b6be : SUCCESS in 7m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5b5a4e133f514e6eac7204d965479710 : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d03f4b162f1449419e77fc510a5e5ea4 : SUCCESS in 4m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3668c02571504aecb292014e14592a50 : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":7},{"id":"e2b20c32e84eb1c5ebe6bb1400d8f03aec629d8a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 01:46:00.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"70cb5ccf5ad0344020bb02265c737d772dd28ec3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-29 12:05:14.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3e621f6f3c2be1d0f1e046d4e224df8faa5711f9","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-29 12:05:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e0d1f2b769991887907958bd8978f7eefd8251a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 13:12:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c084a658ff46858fd52284d6a3a151\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a2444a82ad4337a7412dfbfeb5307a : SUCCESS in 3m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/596e78a6eb5e458ba0003f4d8bd6346f : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/cec7b10990d24f0cb52311990877dee3 : SUCCESS in 4m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1297fb3fe6db46589417fd328134195b : SUCCESS in 4m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/aea3d6d2a0dc46269c798cca1c1c90fa : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":8},{"id":"803253ca797b03899730e77df32db557b98b7ce8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-30 04:04:39.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"b7b25aa1388fe2f888009149e9e2e11c926f0781","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-30 04:04:45.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"6b85ba57e4f9c53a010837f12d114a4092acd020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 04:05:09.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"99026810287f03b5877470eee5960dd528ecf240","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 05:31:17.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04013a1ffdf544f693965961bd12585c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da8ac87938ff4b249021fb331c81fc71 : SUCCESS in 5m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3080edfa00f2436f9b188b8d0c2e52d0 : FAILURE in 1m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/65ebbf70be7b4244b6b668f03a461b6f : SUCCESS in 3m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b1ec476bace14096b1f30bf0bbf37eee : SUCCESS in 4m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3ed1ed39ba2242e090d07a0e0c98a5c5 : SUCCESS in 44m 00s","accounts_in_message":[],"_revision_number":8},{"id":"630ff52a1c5a0a43bf86549477ae8876a58eaffa","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-30 06:34:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"85adb493c960f84b2452a85f0bece8d02a0fea43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 07:32:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c038b4790764fab9e350a54363a0818\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/612b7742e825487c87ff885281c438a0 : SUCCESS in 5m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/03d39ab0fbfb48fcb0a801e8ed41f340 : SUCCESS in 10m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5e38f6d8d808497994bc6084d9f0d1d8 : SUCCESS in 6m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7e7156c7cd814bcd9736a832b18aa9b2 : SUCCESS in 6m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c2543e758be242b7880b4b21a4459c99 : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":8},{"id":"121a0bb0d712863f9399ce5891d7871d0d6ca145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 07:32:58.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f39c85ce5a4cfae6e9d887f12c93837df460700c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 08:32:20.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/636c37f202c04fcfa45f69fb44f72a38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e43cfcd44f4aa691e49efe4ebebd60 : SUCCESS in 6m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/83f2768848ef4d2bbe950bd3c6fdc5db : SUCCESS in 8m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ef1c7c08df77474f8201ea3c0eb4107e : SUCCESS in 4m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b999742ac2744757b1e6165e7ca65586 : SUCCESS in 4m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c46b73c2cbb343a592d0b51676ad07d9 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":8},{"id":"0c7c32d821eb5d75ab73ef0924e40032d5f64cf0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 08:32:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"e527192120736044a1c9eb4d1406d79a479ebf4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 08:33:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9694bb0780c943ca9f8e2520923d77c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a618406f4b6d44688e285275b42e9e2c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"febec304e5554943c42df5807a1e4865dc60849d","revisions":{"a6f0e8c90bfcb531c51e002d55ca9b2485fafd91":{"kind":"REWORK","_number":1,"created":"2025-08-21 12:55:49.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/94/958194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/958194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 12:53:00.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 12:53:37.000000000","tz":330},"subject":"feat: Add disable option for LB member and show provisioning/operating status","message":"feat: Add disable option for LB member and show provisioning/operating status\n\nChange-Id: I83b8b531363aec94634254b35b4ac3b87bd2e8e6\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a6f0e8c90bfcb531c51e002d55ca9b2485fafd91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a6f0e8c90bfcb531c51e002d55ca9b2485fafd91"}]},"branch":"refs/heads/master"},"6081a97ef3b4556260bdff2a285021bb4f49c538":{"kind":"REWORK","_number":2,"created":"2025-08-26 09:51:08.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/94/958194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/958194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/2"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 12:53:00.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-26 09:50:57.000000000","tz":330},"subject":"feat: Add disable option for LB member and show provisioning/operating status","message":"feat: Add disable option for LB member and show provisioning/operating status\n\nChange-Id: I83b8b531363aec94634254b35b4ac3b87bd2e8e6\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6081a97ef3b4556260bdff2a285021bb4f49c538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6081a97ef3b4556260bdff2a285021bb4f49c538"}]},"branch":"refs/heads/master"},"10684cd18c7b9de0fe0ae6c4e1ac75133d059f7b":{"kind":"REWORK","_number":3,"created":"2025-08-26 12:08:58.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/94/958194/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/958194/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/3"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 12:53:00.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-26 12:08:47.000000000","tz":330},"subject":"feat: Add disable option for LB member and show provisioning/operating status","message":"feat: Add disable option for LB member and show provisioning/operating status\n\nChange-Id: I83b8b531363aec94634254b35b4ac3b87bd2e8e6\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/10684cd18c7b9de0fe0ae6c4e1ac75133d059f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/10684cd18c7b9de0fe0ae6c4e1ac75133d059f7b"}]},"branch":"refs/heads/master"},"fbd9aa4f57c24d084ff23b3949b01d6983e9ee87":{"kind":"REWORK","_number":4,"created":"2025-09-15 07:28:37.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/94/958194/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/958194/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/4"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 12:53:00.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:27:49.000000000","tz":330},"subject":"feat: Add disable option for loadbalancer member, show provisioning and operating status","message":"feat: Add disable option for loadbalancer member, show provisioning and operating status\n\nChange-Id: I83b8b531363aec94634254b35b4ac3b87bd2e8e6\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fbd9aa4f57c24d084ff23b3949b01d6983e9ee87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fbd9aa4f57c24d084ff23b3949b01d6983e9ee87"}]},"branch":"refs/heads/master"},"6589aa1523315af0aabb4653d70f0829b2ca81e2":{"kind":"REWORK","_number":5,"created":"2025-09-24 10:47:11.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/94/958194/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/958194/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/5"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 12:53:00.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-24 10:46:51.000000000","tz":330},"subject":"feat: Add enable/disable options for load balancer member and show provisioning/operating status","message":"feat: Add enable/disable options for load balancer member and show provisioning/operating status\n\nChange-Id: I83b8b531363aec94634254b35b4ac3b87bd2e8e6\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6589aa1523315af0aabb4653d70f0829b2ca81e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6589aa1523315af0aabb4653d70f0829b2ca81e2"}]},"branch":"refs/heads/master"},"62c8d5fb4262872c56a0ab2037d2414cbd809196":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-09-24 11:36:40.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/94/958194/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/958194/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/6"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-21 12:53:00.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-24 11:36:25.000000000","tz":330},"subject":"feat: Add enable/disable action for load balancer member and display provisioning/operating status","message":"feat: Add enable/disable action for load balancer member and display provisioning/operating status\n\nChange-Id: I83b8b531363aec94634254b35b4ac3b87bd2e8e6\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/62c8d5fb4262872c56a0ab2037d2414cbd809196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/62c8d5fb4262872c56a0ab2037d2414cbd809196"}]},"branch":"refs/heads/master"},"7e1dcbd6ca120341175eadeba6b2feac779a6943":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-26 10:16:27.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/94/958194/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/958194/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/7"}}},"commit":{"parents":[{"commit":"38fbc9258aaa8f06bbf07c120a336f209c68366f","subject":"fix: invalid vars syntax in playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38fbc9258aaa8f06bbf07c120a336f209c68366f"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-26 10:15:17.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-26 10:15:48.000000000","tz":330},"subject":"feat: Add enable/disable action for load balancer member and display provisioning/operating status","message":"feat: Add enable/disable action for load balancer member and display provisioning/operating status\n\nChange-Id: I83b8b531363aec94634254b35b4ac3b87bd2e8e6\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7e1dcbd6ca120341175eadeba6b2feac779a6943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7e1dcbd6ca120341175eadeba6b2feac779a6943"}]},"branch":"refs/heads/master"},"febec304e5554943c42df5807a1e4865dc60849d":{"kind":"REWORK","_number":8,"created":"2025-09-29 12:05:14.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/94/958194/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/958194/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/958194/8"}}},"commit":{"parents":[{"commit":"38fbc9258aaa8f06bbf07c120a336f209c68366f","subject":"fix: invalid vars syntax in playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38fbc9258aaa8f06bbf07c120a336f209c68366f"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-26 10:15:17.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-29 12:05:00.000000000","tz":330},"subject":"feat: Add enable/disable action for load balancer member and display provisioning/operating status","message":"feat: Add enable/disable action for load balancer member and display provisioning/operating status\n\nChange-Id: I83b8b531363aec94634254b35b4ac3b87bd2e8e6\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/febec304e5554943c42df5807a1e4865dc60849d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/febec304e5554943c42df5807a1e4865dc60849d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
