)]}'
{"id":"openstack%2Fskyline-console~958515","triplet_id":"openstack%2Fskyline-console~master~I33a7c4a4845529c5a6d44060728956701f50dccf","project":"openstack/skyline-console","branch":"master","topic":"barbican","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-09-30 16:51:11.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-09-30 04:14:03.000000000","reason":"removed on reply"},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2025-09-24 08:55:06.000000000","reason":"\u003cGERRIT_ACCOUNT_38294\u003e replied on the change","reason_account":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"}},"38294":{"account":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"last_update":"2025-09-30 16:51:11.000000000","reason":"Change was submitted"},"7353":{"account":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"last_update":"2025-09-30 16:51:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I33a7c4a4845529c5a6d44060728956701f50dccf","subject":"feat:adds barbican panel(named Key-Manager)","status":"MERGED","created":"2025-08-26 07:50:55.000000000","updated":"2025-09-30 16:51:59.000000000","submitted":"2025-09-30 16:51:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958515-barbican","meta_rev_id":"ff7f604f77224cd0efdcaca787d8fa4125b27ea9","_number":958515,"virtual_id_number":958515,"owner":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-30 16:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-09-30 04:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2025-09-30 10:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-09-30 04:14:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2025-09-30 10:52:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-26 07:55:34.000000000","updated_by":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"reviewer":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"state":"REVIEWER"},{"updated":"2025-08-26 07:55:34.000000000","updated_by":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2025-08-26 07:55:34.000000000","updated_by":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-08-26 08:24:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-24 08:55:06.000000000","updated_by":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"reviewer":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"state":"REMOVED"},{"updated":"2025-09-24 08:55:06.000000000","updated_by":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-09-30 10:52:27.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"df40329e774c45ebf4ae0196878facaf4233a919","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-08-26 07:50:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7636936d91cf9e24791d701d413cdde1d26ae196","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-08-26 07:55:56.000000000","message":"Topic set to barbican","accounts_in_message":[],"_revision_number":1},{"id":"995b662541e28e57f530ecaf98e536e8485e2e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 08:24:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd6808b0e9a94c328e9cfdb75d7187be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933bb482358146a6aed71073179c8791 : SUCCESS in 6m 55s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a466c62ace144493be41bd4b46ab6bbf : SUCCESS in 10m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9107c0d1947f4507a7a572bbf0f61899 : SUCCESS in 8m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/91921cd371844aaa9d430784ded67665 : SUCCESS in 4m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f6d774f8aab043658318b4d9ea8e3c7a : RETRY_LIMIT in 6m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/03cb1d8b92c246bbb536634ae6e3305f : RETRY_LIMIT in 4m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d4b156100fb74f2fa06ecff78b137b75 : RETRY_LIMIT in 4m 31s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5c8d2e2f84134a6d963f1bef264f83ab : RETRY_LIMIT in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d66a80433483725c9d183a2b8fd65fe91e90ce86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-08-26 09:27:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ca4eb044125d91c480ec326766b562c40cfe6384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 10:13:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ddf529350014d0d94be4a81a23650bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fdb0687341464da5db5e0a3b7dda45 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4be6cce57e53494db90755cbb2b132bf : SUCCESS in 4m 55s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/cda0dcdbf87a4bf78f2b87d0e4eea15c : SUCCESS in 6m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9e66294ab21f4fbd9da71c3c04ba5a3a : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7f1a4c8f02c04994bcb866353e66af3d : FAILURE in 44m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b0e26a778b514296b31ebefd70b7a013 : FAILURE in 40m 58s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c4ad7428047643aa8539e61bcf9795a2 : FAILURE in 13m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5db7f1d287a04eca844250d33654628b : FAILURE in 32m 58s","accounts_in_message":[],"_revision_number":2},{"id":"f9f4f68b3cc4da133e8080d39d883e9aa34d43e0","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-09-24 08:55:06.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_34517\u003e.","accounts_in_message":[{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"}],"_revision_number":2},{"id":"41ff3110063c65c72ddf8133866c5780727bcea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-09-29 10:20:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f10f375690b1949cbbfec827dc3c4fdbf291181b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 11:48:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c0b3b8bff046448f33043f59ca7c56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b113498fc8647ae8ecf254200f2f51a : SUCCESS in 4m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5736288560c941a7869f306e932a6ae2 : SUCCESS in 10m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0525d2d0dd2d44cf86a82d7203a6bbf2 : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/985187fbc6fe4a6592e335ad4bb6e731 : SUCCESS in 4m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/afe02c1949464f98a1727a10b85fb7b1 : SUCCESS in 1h 22m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ff5aeb6ae6df41a48ae02439f9b62b32 : SUCCESS in 42m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f3515845b5cd4337bd1e0a6a98929211 : SUCCESS in 51m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1795626e53de48d8898ee2faec1184f2 : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":3},{"id":"00bf83b0ceb67e896c8ae77ecd8760e953860fa8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-30 04:14:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e5721f76e434fbb45418e8b297dcd685a6a56571","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-30 04:14:08.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9c1535e302b1442a4405525dd67368b72d5a7733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 04:14:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"80cea3df7d72daa31a575269adaf7ce0a934855a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 05:31:17.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b8da448aa424da7a12c0c574aa881f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1587564d710c4101bbbc6726bccbcf9a : SUCCESS in 3m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2d5a7e94c282416caba6d0025e7ba619 : SUCCESS in 4m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fc41543e8ae64fdeb69a2cd92f37748a : SUCCESS in 4m 51s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d64cfd5b30744de3a1c99e87d31bde00 : SUCCESS in 4m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c2611a7cbaa647daba95824a180e2697 : FAILURE in 52m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9c61f99225634eed8c54482d6d486feb : SUCCESS in 37m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5ad6132425cf46619dbba4527c922e52 : SUCCESS in 27m 20s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bc9885340ec346fca785817328d421c7 : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f1a0a8173f86433688bad2e8a6ece692b9c2ac58","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-30 07:42:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7744df15e32209127d7cbc40dfdd773db27c27ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 08:40:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb73fbc607448a7a3bf69829694d89d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563443d2e4ef41539a888d36e3d2aabf : SUCCESS in 5m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/734c645ae8154bdc8394765954b93946 : SUCCESS in 6m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6330f32bcbda47d0a6ea893d9d08c69e : SUCCESS in 4m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4a40de43d2fd457194e4347c04c2c5cb : SUCCESS in 4m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2833014ed66545598213e20a3c1b73b1 : SUCCESS in 57m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c6d37397ec674034a75e87b50787644c : SUCCESS in 38m 52s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/60f7cc12edcd4c6ca999c9f7a2100bce : SUCCESS in 38m 03s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0ef7d15ab71c4b95bae9cade316deb4e : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":3},{"id":"fdd1d937cc2e5af888e115149be4eb0616d71494","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 08:40:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fd8eb42cf857d315da32758bdeeebc176c27d78d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 10:00:15.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef347c5ffaa2421b8819d643d82f332a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f938dd70f1d43d6bb96926812d8b80d : SUCCESS in 5m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/dc7ba407a1384f34b4146791dcde8873 : SUCCESS in 8m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b4d57ee2a21c43818f6fc16c0c468845 : SUCCESS in 7m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cfb3f73a15a54925b244d7760a93a95d : SUCCESS in 4m 58s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/72c106ccf85443d08f4e7a5b66102fbd : SUCCESS in 1h 18m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bcd39fddad9e42fa98e4616cbf54ab3d : SUCCESS in 51m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2e0f2de2d86b42e589c0fe515dacaee3 : FAILURE in 51m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/66bb264194654a26acdf26939a4b96d4 : SUCCESS in 42m 44s","accounts_in_message":[],"_revision_number":3},{"id":"dff4ef2664086635ba29cab5607d2008b4149f22","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-09-30 10:52:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5448eccc6a4813f6e85216a0ac7efaec9cdbb8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 12:12:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed2627f39cf548ed8819db0e0fd1542f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1425774ab23b42e9ad5df3e10aff6e97 : SUCCESS in 5m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/88bec0c9c5644f0b9f4859f5f271f3c2 : SUCCESS in 10m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c1f87047805848fb8be1cd49e34032a3 : SUCCESS in 3m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0b0ca1b88c3148679266290b70fa1644 : SUCCESS in 3m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/820d8399795449d2a06d99c4a87edcba : SUCCESS in 1h 18m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/57ff3c113fe143bfaa75a53ee57cdfa8 : SUCCESS in 41m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/470a9f100ebe4bcba30f2a9b5c0ffe91 : SUCCESS in 54m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5a9f17d51dd641588ed19c5b6ffc02c7 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":3},{"id":"46fa00cdcbfa5a4eaf1606a3657e58cf28ab199d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 12:13:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2f84fcceee94f8c3ce8a8f058b6aba38329363fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 13:11:49.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a6d5e979ed14110b52a31f0867935d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d14b18d3b78403ea6260c87b112c35a : SUCCESS in 4m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/469f20a7c3a549cea39f8f310fe4f647 : SUCCESS in 8m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/db821fb8b11045feb9fb1c16bbe67441 : SUCCESS in 3m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a5ba5e736df042d09495b320ab798330 : SUCCESS in 3m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a21ef06d27f0473788edc3e3065c37cd : SUCCESS in 47m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9c2a323485394a15b6dd5ef1945e5186 : SUCCESS in 41m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e6d9fec948124981b6c80c6773e8113a : FAILURE in 53m 57s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/44656901f1524df4951af62ebad90a4e : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e62531988b729ad69cdce62782b8fb28c45a3e69","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-09-30 14:24:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a8d4d4ebb21080932fdf26caeb727aaf6654d9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 15:27:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c73c9ef7dcd4baa8fc87eaf2522c40d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011b75623b4a4b9b9984d801933f82e1 : SUCCESS in 6m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a7f92e3fcccf42da949c61b6516a67f7 : SUCCESS in 9m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/29506d6fb6424f098601f1893202dcd0 : SUCCESS in 4m 35s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/be7b5067f55f41beb2a8c6a5c5ff62fd : SUCCESS in 7m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fcff2d4deb5644ebb4bffadc5bceaab5 : SUCCESS in 46m 04s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b71dd50b26f0427aa70e5bd11772bd4b : SUCCESS in 57m 13s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ec992f3c04a04b898ca5e80297976889 : SUCCESS in 37m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/cb5b183cb66d4e8ea85d2cac6ac3a028 : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":3},{"id":"dbe2bc0f3c90836350fffa1e94c406def6150d8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 15:27:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70b22b8a5c1cedce4bcc4e0aca9972fe90956823","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 16:51:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2272d8b13a5b40928a9afafd1b8e3d4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ffed5a54bc4dcabf52155b9956b324 : SUCCESS in 5m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/132bc17fb0b54767b3cfe2cca4492b4f : SUCCESS in 5m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5e2865489fa34c9f9f21b810347969a3 : SUCCESS in 6m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8d7b4fabb95048059d850de1c4c8669c : SUCCESS in 7m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4b17a7baaa1c40cb8dd348514d47d665 : SUCCESS in 1h 18m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7f5d17c30ac44c468611ad06c107bea2 : SUCCESS in 56m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/83cc101e8714430a8f887fa0ba69d825 : SUCCESS in 49m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/82b0d07c6e36408da0bea1e6db19a12a : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":3},{"id":"6bc8db9120b17d1c7b02ac88fdfdc2444a0fbfe0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 16:51:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ff7f604f77224cd0efdcaca787d8fa4125b27ea9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 16:51:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1ec0d2446eb4e95bf6116e2a01509b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e62cb995ab5407fb7f46aaa4b823431 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fa2959a2b6b10f885e9d1ead21be9b353df3352e","revisions":{"5d12f3d2e801ed769ffaa5adc93b1817a94f7a49":{"kind":"REWORK","_number":1,"created":"2025-08-26 07:50:55.000000000","uploader":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"ref":"refs/changes/15/958515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/958515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-08-26 07:35:47.000000000","tz":330},"committer":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-08-26 07:49:24.000000000","tz":330},"subject":"feat:adds barbican panel(named Key-Manager)","message":"feat:adds barbican panel(named Key-Manager)\n\nThis change adds barbican panel for secrets and  list similar to other panels and subpanel. Has Creation , Detail Secret Page included\n\nChange-Id: I33a7c4a4845529c5a6d44060728956701f50dccf\nSigned-off-by: Reet Srivastava \u003creet.srivastava@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d12f3d2e801ed769ffaa5adc93b1817a94f7a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d12f3d2e801ed769ffaa5adc93b1817a94f7a49"}]},"branch":"refs/heads/master"},"09521a5cf952b4f650296bbb993ac4abe7fd39c4":{"kind":"REWORK","_number":2,"created":"2025-08-26 09:27:10.000000000","uploader":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"ref":"refs/changes/15/958515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/958515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/2"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-08-26 07:35:47.000000000","tz":330},"committer":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-08-26 09:26:38.000000000","tz":330},"subject":"feat:adds barbican panel(named Key-Manager)","message":"feat:adds barbican panel(named Key-Manager)\n\nThis change adds barbican panel for secrets and  list similar to other panels and subpanel. Has Creation , Detail Secret Page included\n\nChange-Id: I33a7c4a4845529c5a6d44060728956701f50dccf\nSigned-off-by: Reet Srivastava \u003creet.srivastava@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/09521a5cf952b4f650296bbb993ac4abe7fd39c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/09521a5cf952b4f650296bbb993ac4abe7fd39c4"}]},"branch":"refs/heads/master"},"fa2959a2b6b10f885e9d1ead21be9b353df3352e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-29 10:20:25.000000000","uploader":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"ref":"refs/changes/15/958515/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/958515/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/958515/3"}}},"commit":{"parents":[{"commit":"873174265e8c606726faa470d355482982aa1840","subject":"Merge \"feat: Add Unrescue feature to nova instance actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/873174265e8c606726faa470d355482982aa1840"}]}],"author":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-08-26 07:35:47.000000000","tz":330},"committer":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-09-29 10:20:25.000000000","tz":0},"subject":"feat:adds barbican panel(named Key-Manager)","message":"feat:adds barbican panel(named Key-Manager)\n\nThis change adds barbican panel for secrets and  list similar to other panels and subpanel. Has Creation , Detail Secret Page included\n\nChange-Id: I33a7c4a4845529c5a6d44060728956701f50dccf\nSigned-off-by: Reet Srivastava \u003creet.srivastava@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fa2959a2b6b10f885e9d1ead21be9b353df3352e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fa2959a2b6b10f885e9d1ead21be9b353df3352e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
