)]}'
{"id":"openstack%2Fskyline-console~958555","triplet_id":"openstack%2Fskyline-console~master~Id09331027e6919643d1dcdc9c32de917cba84d83","project":"openstack/skyline-console","branch":"master","topic":"volumetype","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-09-28 01:54:50.000000000","reason":"removed on reply"},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2025-09-28 02:40:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id09331027e6919643d1dcdc9c32de917cba84d83","subject":"feat: show IOPS for volume type during volume creation and hide distro icons when no images are available","status":"MERGED","created":"2025-08-26 16:02:09.000000000","updated":"2025-09-28 02:40:59.000000000","submitted":"2025-09-28 02:40:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958555-volumetype","meta_rev_id":"4ff5e57ec0f5b6fe379f7b821a8d558349996750","_number":958555,"virtual_id_number":958555,"owner":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-28 02:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-09-28 01:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-09-28 01:54:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-26 16:28:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 03:10:57.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"976837984420c5a8670b354801fa65a1ad2c3346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-26 16:02:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e90c2e6f1dc9c6441f9e90d84a56711247bfe91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:28:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c900f3b933ef4630889f48711510af72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ab63acfe94445a915a5e341e935c98 : SUCCESS in 4m 43s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7457de876f2d43339cda157c4b35601f : SUCCESS in 7m 05s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/da5829cd3e99404f87614c716c244700 : SUCCESS in 7m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c1696cb048b246d2b8887c428e77ced2 : SUCCESS in 4m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/800a950d47e4463a82f5fe9dc2747e42 : RETRY_LIMIT in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ec0e214676b341e48cb16a308480deed85bcafa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-08-29 07:45:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ed25c0a6e9eaee05b1dfdab816c4e0a6cfac6b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 08:27:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b110ce9f53e434781c65a3fc57b97d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35021968d73a4359a2609554d2dd7389 : SUCCESS in 5m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/15bd111d6dcc4960a382b6da3f463101 : SUCCESS in 10m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f7924626b6eb40b493ce842ade3d266b : SUCCESS in 3m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cb70e627e9ac4690a33e5eeda2d3eb1b : SUCCESS in 8m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c6da1531b1564a4eb23f0f0a81c4afaa : RETRY_LIMIT in 7m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e2c9c397b71045a5bc5283fa25e695ec : RETRY_LIMIT in 4m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/49aa29d31f4f4c2085f2627bab4c8260 : RETRY_LIMIT in 9m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c2dd495d32a64c78b3a36145b1fea593 : RETRY_LIMIT in 11m 03s","accounts_in_message":[],"_revision_number":2},{"id":"31c01353e593fcbeac0079dcc019c29b7f2f99ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 07:24:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c7bc85ec4874e93f5ca305be7dd9593661a13f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 07:58:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/724096494048494bb767baa220ebe149\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa5cec72ebd4678883375e2e4c82516 : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/89a1455905a44046ae19bbc94e2a7ece : SUCCESS in 11m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8129f159942c4db1b3a5671ef277861d : SUCCESS in 5m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/80f14e7947894041a0310798bb89f914 : SUCCESS in 4m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/98c75ae60a4442b7ac96d9ff7fb235f9 : RETRY_LIMIT in 6m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f79acb72b0c4984c0f2f05f87e2ff1574fe8bd2e","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-23 03:10:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fbc79c9b7ebda86e183158718f5ec75a26299e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-23 07:45:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"84e68506f447a541c91752be2a799954004b3941","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-23 07:46:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"53a6dc1faa00d0cec203ecc8b91938c3291480f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 08:08:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d18eb1ac06fa415291edb269acc52f38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fde265545ab475d95b4f94cc6998f26 : SUCCESS in 4m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3b50f478fca74378a9a896dcb7f46577 : SUCCESS in 10m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/84ae935ea6a94f1bbebedd128bff6b95 : SUCCESS in 7m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/23b2e59fe88246278a6ecb507e454a0a : SUCCESS in 4m 07s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a2c3252f3acc4cf7b8cc9320d0b0bd5e : RETRY_LIMIT in 9m 31s","accounts_in_message":[],"_revision_number":4},{"id":"d4eb17b4921a3a7670a33c7ffcf8229d703901a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-26 13:05:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"79e04b7d4712da4c403fe1482f654b4b1b1abf7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 14:01:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c7d488efb174e708b96f76ae77ba23b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02ead62ea764739a5d41a46691865f9 : SUCCESS in 5m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/60d622ae9ded4f5d86cf9a0cdc980809 : SUCCESS in 10m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3cf31717202742a69db92896477f483a : SUCCESS in 8m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/11e5a20e65ee497e910bbda88972ced1 : SUCCESS in 8m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/023b7ce5b45040079e238404896f7e80 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":5},{"id":"3df3cd2c7ccc70abcac305e5498231c0b452704b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 01:54:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"40a463b457596526486b54f87f78ebb0fd9a09d7","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 01:54:58.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"30992a4ee5b8e26a73ea5569b0f3756d78b0041b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 01:55:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cd8e8e6adb005d7f4397e44ade9a22f4ff969c95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 02:40:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcfd9e617b234bf9869bcf2390350673\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a2c0e725fb4ad79c7141790722d77f : SUCCESS in 4m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1ce1fa49b57d436b8f5ff54d7bf89bde : SUCCESS in 4m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3dfb49df63c04e679bda92ad0857cc0d : SUCCESS in 4m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/35693ae84cfc47c2b07b26f65a19cf67 : SUCCESS in 7m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/986e427cf6fd439c9e7419e65afbefd1 : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":5},{"id":"9e9d85103ae180ab17a309c55c85abee719be072","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 02:40:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"4ff5e57ec0f5b6fe379f7b821a8d558349996750","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 02:40:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e709695a2b44231bfecfe0316f8b81c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d7472754204067b44d85cdcf39791c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"81fd878732c36acee2898733fa3abe55380ef99a","revisions":{"6f5298c9f0d73f7d4bef3599f97a671d61c2a95e":{"kind":"REWORK","_number":1,"created":"2025-08-26 16:02:09.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/55/958555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/958555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-26 11:33:41.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-26 15:55:13.000000000","tz":330},"subject":"feat: show IOPS for volume type during volume creation and hide empty distro icons","message":"feat: show IOPS for volume type during volume creation and hide empty distro icons\n\nChange-Id: Id09331027e6919643d1dcdc9c32de917cba84d83\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6f5298c9f0d73f7d4bef3599f97a671d61c2a95e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6f5298c9f0d73f7d4bef3599f97a671d61c2a95e"}]},"branch":"refs/heads/master"},"d2137b2a82d9f62518ffcaef65fa066a03deaf0f":{"kind":"REWORK","_number":2,"created":"2025-08-29 07:45:00.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/55/958555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/958555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/2"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-26 11:33:41.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-29 07:44:27.000000000","tz":330},"subject":"feat: show IOPS for volume type during volume creation and hide empty distro icons","message":"feat: show IOPS for volume type during volume creation and hide empty distro icons\n\nChange-Id: Id09331027e6919643d1dcdc9c32de917cba84d83\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d2137b2a82d9f62518ffcaef65fa066a03deaf0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d2137b2a82d9f62518ffcaef65fa066a03deaf0f"}]},"branch":"refs/heads/master"},"a4040abe6bcea8ad46651e62184137f771ee0a76":{"kind":"REWORK","_number":3,"created":"2025-09-15 07:24:50.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/55/958555/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/958555/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/3"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-26 11:33:41.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:23:10.000000000","tz":330},"subject":"feat: show IOPS for volume type during volume creation and hide distro icons when no images are available","message":"feat: show IOPS for volume type during volume creation and hide distro icons when no images are available\n\nChange-Id: Id09331027e6919643d1dcdc9c32de917cba84d83\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a4040abe6bcea8ad46651e62184137f771ee0a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a4040abe6bcea8ad46651e62184137f771ee0a76"}]},"branch":"refs/heads/master"},"9da4e69c9ae2b15456c66c720f8f23c3513424c4":{"kind":"REWORK","_number":4,"created":"2025-09-23 07:45:00.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/55/958555/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/958555/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/4"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-08-26 11:33:41.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-23 07:44:44.000000000","tz":330},"subject":"feat: show IOPS for volume type during volume creation and hide distro icons when no images are available","message":"feat: show IOPS for volume type during volume creation and hide distro icons when no images are available\n\nChange-Id: Id09331027e6919643d1dcdc9c32de917cba84d83\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9da4e69c9ae2b15456c66c720f8f23c3513424c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9da4e69c9ae2b15456c66c720f8f23c3513424c4"}]},"branch":"refs/heads/master"},"81fd878732c36acee2898733fa3abe55380ef99a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-09-26 13:05:16.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/55/958555/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/958555/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/958555/5"}}},"commit":{"parents":[{"commit":"38fbc9258aaa8f06bbf07c120a336f209c68366f","subject":"fix: invalid vars syntax in playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38fbc9258aaa8f06bbf07c120a336f209c68366f"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-26 13:04:48.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-26 13:05:00.000000000","tz":330},"subject":"feat: show IOPS for volume type during volume creation and hide distro icons when no images are available","message":"feat: show IOPS for volume type during volume creation and hide distro icons when no images are available\n\nChange-Id: Id09331027e6919643d1dcdc9c32de917cba84d83\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/81fd878732c36acee2898733fa3abe55380ef99a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/81fd878732c36acee2898733fa3abe55380ef99a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
