)]}'
{"id":"openstack%2Fskyline-console~961092","triplet_id":"openstack%2Fskyline-console~master~Id7c86d1093314f6e4e8b9dd8b106df883b671ce0","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-09-26 02:09:44.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-09-25 14:10:44.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2025-09-26 09:10:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7c86d1093314f6e4e8b9dd8b106df883b671ce0","subject":"fix: invalid vars syntax in playbooks","status":"MERGED","created":"2025-09-15 07:01:30.000000000","updated":"2025-09-26 09:11:18.000000000","submitted":"2025-09-26 09:10:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961092","meta_rev_id":"b76fe73091d2574c7c82e691aa5df26d591f8669","_number":961092,"virtual_id_number":961092,"owner":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-26 09:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2025-09-17 23:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-09-26 02:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-25 17:44:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2025-09-26 02:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}],"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-15 08:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-17 23:48:33.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-09-18 00:04:18.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"CC"},{"updated":"2025-09-24 07:21:12.000000000","updated_by":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-09-25 17:44:50.000000000","updated_by":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"reviewer":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"state":"REVIEWER"},{"updated":"2025-09-26 02:09:04.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"3a26c7069c09272ddb4763e168120e5275b7ef7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 07:01:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf051cdf2d854125f07174044c5315ff2db7f39e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 07:09:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7c826bb1bb2df0cd795bb5bb216d32ff08a6fc5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 07:13:21.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b574dda1203ab00e83e974b7bf6c59a5303434e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 08:27:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ab8940bf15043288d5a3c9ee5a3608d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e233e30c21b4040a05ad64e5e3adc5f : SUCCESS in 6m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/531a3d3da2ec46e5b132153a4b0f38f5 : SUCCESS in 10m 24s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a0ce4c3e2d984396a02ddb4e491e8451 : SUCCESS in 5m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/392a455e180c4f94ade350ad3d88cf3d : SUCCESS in 4m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6576e6eca1c04ff5a579fb46700a2544 : SUCCESS in 1h 08m 50s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bfbf32058852421b90ad2715a2ce3db9 : SUCCESS in 59m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a9399649541e467c85b0c25392458831 : FAILURE in 1h 03m 45s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a2b85d0ead02451eb99e4c991822425a : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":3},{"id":"ed0d8c811edf5ad7f9a76a1b7a507c14dc3a329f","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 08:50:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"30495bf1dfa65a1ee236725f6eb7e1da8abffaee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-15 08:51:17.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b4cf22460f307d0e2eba46f24c3711d8372dba79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:19:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbf7daaf3e2049b980414c986865f912\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/511fc01caacc4265abf6179ddf29196b : SUCCESS in 5m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/80ecc7c0c01241f284a1e27461ab753f : SUCCESS in 10m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3d88c5e8d294423eaba0a2c039aaaf8b : SUCCESS in 4m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3451f7400e704ecf85242c16288f452b : SUCCESS in 4m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2739570f7b7349799e257358bbc82af0 : SUCCESS in 1h 08m 05s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/53532c42fcfd4d58bb4d85f433cc25e6 : SUCCESS in 44m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8fdbc0e690bb457f8d6f5c25b39ad3c9 : SUCCESS in 53m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/fd817e4469ac450783302fe4af47fe9c : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":4},{"id":"cfa846f1bb6e21befed2740f5c0e68e558ace438","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-09-17 23:48:33.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f1f40fab179fe267405d87725351f5c555c324a","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-09-18 00:04:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2f6223404542a620ad4a7d2a7680da454a3dd0d9","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-18 07:01:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3db1ab493b9f31e79ce02820a3ed27e7ab966093","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-09-25 14:10:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c8e4c82f757e0c9e484238b2faac95c499de3afd","author":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"date":"2025-09-25 17:44:50.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cce9e7283962537d1ac713f90466fe03da3debc0","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-09-26 02:09:04.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bb2743d5045626921626899131291cd39e8674bd","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-26 02:09:44.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ec5d010170bf1ac419fb257012ed4eaf74d1c366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 02:11:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ae6d3e6fab23d0e5dc2c6ccc2960d5f52f4f13eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 03:05:31.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a71a1edca5d4807b70c2f548aeb00f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55578a46a39b46a9b446831d9e1b8536 : SUCCESS in 4m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f33af8fb7d824af697102507c2112f1c : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d5d1928f8caa431ebaa22174c05344a2 : SUCCESS in 7m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/395c857c05d54cec91a1202d045685de : SUCCESS in 5m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c0b9a253f1924539a16901b3b5aeb3c4 : FAILURE in 52m 26s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b6c7d23182194c2d8f7ab9a48f217918 : SUCCESS in 30m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/91df420e8217441f8bfedc213c8d4911 : SUCCESS in 49m 26s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/acf55e2ba9254577b145c62c7b70f484 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":4},{"id":"b6e2ab0100215e5fdbe5c579db17e2ce5e83d679","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-09-26 03:44:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52d18857f61b711f26e1b6e25a09688acb448182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 05:06:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c39de0e5c1ed4b9c9046655709d80c6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/910f8c4518d946de9ba743fba61b9f94 : SUCCESS in 5m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3057a1a4582343df8c0eca071f9dd530 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6dc9e6145c594b1f954f23514d5bf58c : SUCCESS in 5m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/178cf5c29f004b29979535b230d0ee8f : SUCCESS in 3m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fbd71e2f1372429ab6db494b1044ee2b : SUCCESS in 1h 21m 39s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/22b3c716598a4e9aacfc55d7770f95c8 : SUCCESS in 34m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/47418c65ca5d4b01b3de67ca8be4b9da : FAILURE in 56m 37s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/463258d02b33489cade9364db39ecec9 : SUCCESS in 42m 24s","accounts_in_message":[],"_revision_number":4},{"id":"aa4fb5d4d741e436ee7edb6ffcb2662c7c5c135f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-26 06:31:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a38c59c3039743e7fb013691fda0b4bd9335d437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 07:55:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca1081ee60764d5894bffbf42e67e1d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa2e2a5dc1b421ca491755c9ecec9a6 : SUCCESS in 6m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4deb5b211db643fbb9ab4973bcf9dfcb : SUCCESS in 10m 24s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a1d2f1a570d240758b77ecfbbaa38be3 : SUCCESS in 7m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/def9473b909740829604cecf727fa902 : SUCCESS in 7m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3a6f59a26c4946858dc839d2adec0259 : SUCCESS in 1h 23m 05s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5fc24b8f69e44c3282429c11244d001b : SUCCESS in 59m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fb9f796cd741460c8292dfef0ee422d5 : SUCCESS in 52m 10s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f905c62bdb3447f0951ddda49b064a18 : SUCCESS in 49m 31s","accounts_in_message":[],"_revision_number":4},{"id":"0435965405a32d84d51676fbdb8486790fe9a946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 07:55:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b8f97f9fd3969995e0f40bdf67d6ca3e0530b0a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 09:10:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b171201d7b4686b8734a0f61d72f35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9001f6fda19d4ab1b9e1c252e2800bfa : SUCCESS in 5m 51s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b2a3369b65f24f6082c78720dd0112f5 : SUCCESS in 4m 02s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2adfffe1df734b6e9f6d1453fbacee29 : SUCCESS in 7m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8047cd51b3ae47fe89ff945120e6c284 : SUCCESS in 5m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7b34e8136f5146d0a236ea9e4386ee99 : SUCCESS in 1h 13m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/28ac27e8f5de43cebab3fd804318dc4c : SUCCESS in 50m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/68f4b95509ee4bd4829bb8688db2e93b : SUCCESS in 46m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/dd4212d057614551bb8f1d7fdb9d3c57 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":4},{"id":"59700845cb0ea9f1e6af2f9256a8028d8200aac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 09:10:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b76fe73091d2574c7c82e691aa5df26d591f8669","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 09:11:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eaf273bbd9e404ba89a51943379dffd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23038ac0596e4449a3bdea3dac05512f : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"38fbc9258aaa8f06bbf07c120a336f209c68366f","revisions":{"6e5efc576aa92f76c8fbb2b396cee4623cde6ff6":{"kind":"REWORK","_number":1,"created":"2025-09-15 07:01:30.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/92/961092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/92/961092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:00:08.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:01:19.000000000","tz":330},"subject":"fix: correct invalid vars syntax in playbooks","message":"fix: correct invalid vars syntax in playbooks\n\nChange-Id: Id7c86d1093314f6e4e8b9dd8b106df883b671ce0\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6e5efc576aa92f76c8fbb2b396cee4623cde6ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6e5efc576aa92f76c8fbb2b396cee4623cde6ff6"}]},"branch":"refs/heads/master"},"03eb97978a8222b210c6948fa47e3ed1591f4d55":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-15 07:09:09.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/92/961092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/92/961092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/2"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:00:08.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:09:09.000000000","tz":0},"subject":"fix: invalid vars syntax in playbooks","message":"fix: invalid vars syntax in playbooks\n\nChange-Id: Id7c86d1093314f6e4e8b9dd8b106df883b671ce0\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/03eb97978a8222b210c6948fa47e3ed1591f4d55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/03eb97978a8222b210c6948fa47e3ed1591f4d55"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5098e2108b87ae43159c100e211c55483773cf76":{"kind":"NO_CHANGE","_number":3,"created":"2025-09-15 07:13:21.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/92/961092/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/92/961092/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/3"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:00:08.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:13:03.000000000","tz":330},"subject":"fix: invalid vars syntax in playbooks","message":"fix: invalid vars syntax in playbooks\n\nChange-Id: Id7c86d1093314f6e4e8b9dd8b106df883b671ce0\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5098e2108b87ae43159c100e211c55483773cf76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5098e2108b87ae43159c100e211c55483773cf76"}]},"branch":"refs/heads/master"},"38fbc9258aaa8f06bbf07c120a336f209c68366f":{"kind":"NO_CHANGE","_number":4,"created":"2025-09-15 08:51:17.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/92/961092/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/92/961092/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/92/961092/4"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:00:08.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 08:50:56.000000000","tz":330},"subject":"fix: invalid vars syntax in playbooks","message":"fix: invalid vars syntax in playbooks\n\nChange-Id: Id7c86d1093314f6e4e8b9dd8b106df883b671ce0\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38fbc9258aaa8f06bbf07c120a336f209c68366f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38fbc9258aaa8f06bbf07c120a336f209c68366f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
