)]}'
{"id":"openstack%2Fskyline-console~961168","triplet_id":"openstack%2Fskyline-console~stable%2F2025.2~Iae3db9314d62be710a697ce004d8cac9ec1acc4b","project":"openstack/skyline-console","branch":"stable/2025.2","topic":"create-2025.2","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2025-10-11 10:03:32.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-10-11 05:49:55.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-10-11 10:03:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae3db9314d62be710a697ce004d8cac9ec1acc4b","subject":"Update .gitreview for stable/2025.2","status":"MERGED","created":"2025-09-15 09:56:21.000000000","updated":"2025-10-11 10:04:16.000000000","submitted":"2025-10-11 10:03:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961168-create-2025.2","meta_rev_id":"736d8bf926192063b9111cd370c5e0b801dea555","_number":961168,"virtual_id_number":961168,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-11 10:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-10-11 05:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2025-10-11 01:28:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-10-11 05:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"CC":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-15 10:33:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 02:16:42.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-09-30 22:24:20.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2025-10-04 00:40:21.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"},{"updated":"2025-10-09 03:32:02.000000000","updated_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"55b26b50dbb74ee10ec4270054dd4974a36114a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-09-15 09:56:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18658953f96f7b1b8ffe9740db091db9050fca4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:33:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf636c02e604cc8890a115df13e4dde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3866ec3a3d248749258af6938cecdc4 : SUCCESS in 7m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/927958320d054b4d885c85563a5cbc06 : SUCCESS in 9m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/709fd948e4834643a71278902fa1c72f : SUCCESS in 4m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f52b079d97f840b99699f0db4fba82ed : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0fa07da48c091b054b09a3f558efcc980665bc24","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-23 02:16:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9925c38a4e35fc5df481e50fc2b9965c2128e1a9","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-09-30 22:24:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5d5422263e245a5b9259e70523e3cf150850ffd","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-09-30 22:24:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68a49f931165723db243dea540ca6d3c86702909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 22:25:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1494ccd7bf002d7f983414caf5a0476011c2e8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 23:07:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a75c21bd07b0427f8cf0caab22c903e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f9cb12dbbc4e74a9aa7a1886769157 : SUCCESS in 5m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/526b6fa2564b471eab279157eac68f27 : SUCCESS in 4m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/80f188bcbe904bf5a93cea25922f9857 : SUCCESS in 3m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0c11081be3104fd89b2f83e41f164ddd : SUCCESS in 6m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/09d9d0530ceb41b5a15bc7c843829095 : RETRY_LIMIT in 6m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/162a72fef5124c55aba9fa20d0368d66 : RETRY_LIMIT in 3m 57s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5e82320b6b084fc9904033fe3660fc85 : RETRY_LIMIT in 3m 48s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e6f1c2aa52624cf6b78c67bb1babb3e9 : RETRY_LIMIT in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"88a0746703c3db4311fe257a33090792c2ef7e0b","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-10-02 11:36:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fda8ec0d39aafde9f012f88bc2f44829424c1e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 12:33:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79a02d41d969450fb2200c61347a9d8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252ce555ed854f1cbecd9fa356bd070a : SUCCESS in 5m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0da0ee81c1084a47bdec7c45e346c838 : SUCCESS in 8m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/10195c7cc3c141919803c08c92dff574 : SUCCESS in 10m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6cf8dab703f84c36b618a8adb0ca9eef : SUCCESS in 4m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/46ea959ab897437c857daea276cdc5b4 : RETRY_LIMIT in 11m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/580668a1b89d4a8cbdbcd6f7f4768ecc : RETRY_LIMIT in 4m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4cf323aa65734999855070b065652ad1 : RETRY_LIMIT in 10m 21s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4601283acae94a818ef03192789a08bb : RETRY_LIMIT in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"92551ffe062779b9a22984b2ddf7e9ba3ce18b5d","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-10-02 12:36:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"955c34e4ff2d61a99905951efa4e8ef3d65b4041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 13:39:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd8b93574d1d4ebe89e1efe7aefcb733\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec740a7be4444f2a286a104ce662447 : SUCCESS in 7m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/06d462eb6d2a42478c2a2124c7a0bda1 : SUCCESS in 11m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b93bdb4dfe9f4b49ad89216ad5c2e461 : SUCCESS in 5m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6c89f875264b4f9a97fea6dc8e61b5df : SUCCESS in 3m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1a1741a2ce14409c9ffd44c34d686cb1 : RETRY_LIMIT in 7m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a9714801cc8e418e90a3b8ebc536a0aa : RETRY_LIMIT in 10m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bffe4afcab6e41dab0814f181424d818 : RETRY_LIMIT in 8m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5c6b8ef4d540478d84f3d4a974c4e092 : RETRY_LIMIT in 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"21606d40b28585e97d7dfc56c3f71273a3960806","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-10-04 00:40:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"724842cd08489bfc641ad1deebc9f6cf3582b50c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 00:48:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aacd17ab26854c32ac05b80d063a97ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7baaccc0e7224851af3fc8860e799955 : NODE_FAILURE in 0s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ae100a080279492c829dc9d5bb9b0098 : NODE_FAILURE in 0s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/90bb41a0fa1540bb964ee0e1b539c088 : NODE_FAILURE in 0s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/279c18e6743440b68395d22966fd3483 : NODE_FAILURE in 0s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/28830dbf3f6a4f7bb3fccce40097605e : NODE_FAILURE in 0s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/260234b5c8704f62aad14c4acaa4fa11 : NODE_FAILURE in 0s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/88d2de3414db445f8f5c16fd4272a1ff : NODE_FAILURE in 0s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/23007cbde14c4b5cbc81b4114e3b4cb0 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ff5e7510ca4e8dd385da12ca496108b72e953f9c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-07 05:13:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a957ed2b65fbedd27f1148a026dee5305a4ae15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 05:52:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcdc764a2fd240c2884784516b897172\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5046991f1bbc4117b3c75ebc4597231c : SUCCESS in 5m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5b9a46315a7e4e8a958c50b4d3c076aa : SUCCESS in 5m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/764b43bcafde4aabb7d04695ee0ae426 : SUCCESS in 6m 58s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/24ae25f0176249a2836c1549b9dea85b : SUCCESS in 6m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5e8c957599674f659d1c6ba342e2a996 : RETRY_LIMIT in 11m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/48218c7a2bfa4b6d9ecec06cf74d1a54 : RETRY_LIMIT in 9m 55s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fe40870290bb44158fe25a7769cbc40f : RETRY_LIMIT in 10m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/373213d62f4c45fda1163911d20f3376 : RETRY_LIMIT in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0335f35cea72eb5227a886c8d51fd03644befeb3","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-10-09 00:48:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a1bddf5b330bdf200bf2d1d2d7b76e8f00bc5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 01:20:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a6ac6ae8a99456db6c70c1a6b988cff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7467e3e2a74d019437b50f4c42c184 : SUCCESS in 4m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f208a7f23c1b48b39761ba37a5136901 : SUCCESS in 11m 55s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/957d960b1aea4d1cb04d02b3bf339e90 : SUCCESS in 3m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4ac0b7a4bc6b439285b0d3aeb79051f7 : SUCCESS in 7m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/77007e84fb2045fca88613dd30ee7d27 : RETRY_LIMIT in 5m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/508fbac7212f4e4b9fc7affb07126029 : RETRY_LIMIT in 5m 50s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5f135c452f8c47e6930276d76a093ef1 : RETRY_LIMIT in 3m 58s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b3c1c7e535a54c83bce003ecbf55263a : RETRY_LIMIT in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"88086966d138f1c206e719842418f1783086d6c7","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-10-09 03:32:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fe90a43aad7057b4b822306303558a460a39223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 04:10:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06e194aad8204074b5a86417ea2e75c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf2e2c9fe0b44882ac346cd377548906 : SUCCESS in 3m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3a9e3491251d45c2bfa169bf65a43583 : SUCCESS in 11m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/81f1251246e74c2289d8eadb705df7e0 : SUCCESS in 3m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/96c39ba8ff834a4c9ac9ada6d15af80b : SUCCESS in 3m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ee0a4b60319c41e38acec1406de9a614 : RETRY_LIMIT in 5m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/704e98c126094dc79639c02a7c74076a : RETRY_LIMIT in 4m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e1ea47bb44654fa78920de4b5f340dc5 : RETRY_LIMIT in 4m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b8bee0e7d09649bbbbb9a75320eafef6 : RETRY_LIMIT in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2400b3179596af4a8ce05d085d6a5c18cdb20a3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-11 01:28:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22816\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"9357353c8de903f5d317a553996c2901ab74c800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 02:25:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7a2258bc5f149eb8128b61526c152cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc34ea9b1264f2cb2d96a6806b88347 : SUCCESS in 5m 55s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e1d84d7f55614ed4bb41d422cee9a2b3 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6153e7ce918a44d5b104aa5ae62c66aa : SUCCESS in 4m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9b5cfab74a454e6fad6783ecdaf7866d : SUCCESS in 5m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bfb276d63b674abbb0c0be2b6126d7c2 : FAILURE in 55m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9f43d2a562bc49959f28a447fc43a27e : SUCCESS in 34m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a623a2d7e5514ddc92febdb603d8394d : SUCCESS in 52m 12s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9838fa465f274c0497503070bdfdac3e : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3cca225a6a5668b49d9e5d825ba030c4c2f1da4d","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-10-11 02:55:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"502abd7e933f3f60a780040f6b2ea4510df3c956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 04:22:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/832f2f68c6074862bd3bc22bc569a05a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/192927c28d444b0b82299219dbeb7171 : SUCCESS in 5m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/85390e2e9d8c4917b4387b9760cb757a : SUCCESS in 9m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f0ee8cd7689d4ec5bdc699c6faa30e2c : SUCCESS in 10m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/70d47a1938654b909790f3aa7c78796e : SUCCESS in 7m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/924066433417482593ea2d3abead63ec : SUCCESS in 1h 21m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/00bcce6239d540b2aca0ea0187fcbb75 : SUCCESS in 1h 09m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3c37fe661edd46d39e3a364aa73eb62f : SUCCESS in 50m 35s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0bf519bb1c79422396683b8c47845cf2 : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":2},{"id":"cfece519a7c2cc9f52245f0fd62b360790a0603f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-11 05:49:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"07bcca1704faf612f28926ffc501f306d9ea65af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 05:50:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ae2ecbb74c64a1194f06b7d72f3097d18645eab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 06:55:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/969fd93df7074ff79aca8d7473c6be35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d89b52dbe74383aedca44bab002727 : SUCCESS in 5m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5eb2823fe23c4f3da6f65ead92b9c321 : SUCCESS in 4m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f3110da1f3e54ae297ccbec675419a7c : SUCCESS in 6m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9965a816c54c447f9e168d58d8ad6859 : SUCCESS in 5m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/788cb4f92d034e97adf003d15792588f : FAILURE in 50m 12s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/66202853035547f0abc932c274bafb0d : SUCCESS in 1h 00m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d1e88cdf1a5d45ee92339d7443c05c15 : SUCCESS in 51m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bbb22b2ba93246ae91da1042ba616869 : SUCCESS in 51m 57s","accounts_in_message":[],"_revision_number":2},{"id":"ac529eedf51d6e0a0b4e8293bb6fd2c744b03fbe","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-11 07:42:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd4490843b470e3116cddfd2501ee09a06758439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 08:43:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dc502208c554973a0eac0f613b07b55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c20ac0698a4d369528c9b9415ae505 : SUCCESS in 3m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/76574de8b0fb47b3982c96f829bb7d16 : SUCCESS in 6m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d458a757944c41d9a794b8193470479d : SUCCESS in 3m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/97ac1a2839014528bb87e433f867a85c : SUCCESS in 4m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/47146708ac1e4746b13806231096c750 : SUCCESS in 1h 00m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2ed2855598934ce3b85055fc08ef1a3c : SUCCESS in 41m 55s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6e3a59580e5b497faaf14ce86d693c0b : SUCCESS in 50m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7e9cf708675d4a86a6b21d7f5a8a90cd : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":2},{"id":"542d69651a84b3c59dbab5b4b9d9d93ad975ebfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 08:43:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a865a197574bc36bbd49f066ab5fd6dc0d1e843e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 10:03:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/862d95980bb44ccbbc4e55a1ddd0f094\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bfb8ebef35d4006ab249764f2b5bc30 : SUCCESS in 5m 42s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/13bae83bf49943899996f5a4694504b3 : SUCCESS in 6m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7df395d97b074f71ac1207f0512ea008 : SUCCESS in 3m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/dfa86bcae8294c8ebefa03f5851d2214 : SUCCESS in 7m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f41b5aa912d54469b77313ec9f501015 : SUCCESS in 1h 19m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/cfcd96327a054274a40014bde6e3026d : SUCCESS in 1h 00m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/08e5d9060b2d48d49aafb70b09ba4765 : SUCCESS in 53m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d0fdcfe57fb74badb85cbef55383b501 : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d2bc7c1143ac0b0fae36813ac8ea8dd37fa59325","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 10:03:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"736d8bf926192063b9111cd370c5e0b801dea555","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 10:04:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d89e9a219a4089b96f1311d96a7551\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8291a6acb4b74fd6ab348d57e04a01a3 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f1e8146b63e5d33ad9896279b34feef21d4d523a","revisions":{"64309e3f6ed0088a3e627bf2aec634105b7554d9":{"kind":"REWORK","_number":1,"created":"2025-09-15 09:56:21.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/68/961168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/961168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/961168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/961168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/961168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/961168/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-15 09:56:20.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-15 09:56:20.000000000","tz":0},"subject":"Update .gitreview for stable/2025.2","message":"Update .gitreview for stable/2025.2\n\nChange-Id: Iae3db9314d62be710a697ce004d8cac9ec1acc4b\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/64309e3f6ed0088a3e627bf2aec634105b7554d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/64309e3f6ed0088a3e627bf2aec634105b7554d9"}]},"branch":"refs/heads/stable/2025.2"},"f1e8146b63e5d33ad9896279b34feef21d4d523a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-11 01:28:17.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/68/961168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/961168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/961168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/961168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/961168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/961168/2"}}},"commit":{"parents":[{"commit":"67fa5ee2b28e718168017d33e86fc9eebcc8a7f9","subject":"fix: invalid vars syntax in playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/67fa5ee2b28e718168017d33e86fc9eebcc8a7f9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-15 09:56:20.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-10-11 01:28:17.000000000","tz":0},"subject":"Update .gitreview for stable/2025.2","message":"Update .gitreview for stable/2025.2\n\nChange-Id: Iae3db9314d62be710a697ce004d8cac9ec1acc4b\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f1e8146b63e5d33ad9896279b34feef21d4d523a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f1e8146b63e5d33ad9896279b34feef21d4d523a"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
