)]}'
{"id":"openstack%2Fskyline-console~961359","triplet_id":"openstack%2Fskyline-console~master~I64be95aa88fd1e4ed063bbbaef86b3696862b50a","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-09-28 02:02:18.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"38294":{"account":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"last_update":"2025-09-28 05:44:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I64be95aa88fd1e4ed063bbbaef86b3696862b50a","subject":"feat: Add Unrescue feature to nova instance actions","status":"MERGED","created":"2025-09-16 10:07:29.000000000","updated":"2025-09-28 05:45:46.000000000","submitted":"2025-09-28 05:44:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961359","meta_rev_id":"cab012c712c54506cfb9e6b32ee1c77f8c285f69","_number":961359,"virtual_id_number":961359,"owner":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-28 05:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-09-28 03:55:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-09-28 03:55:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-16 10:54:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 02:37:45.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"df8bf1e0fea2f9b35fdc4b5769c754572b7531fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-09-16 10:07:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a4f347d271f8e8d3f6d8b4b08324b6064ea37f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 10:54:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abd258391b524fe69ea0fd62d5011e0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ddab10759a487db407398bbab045a4 : SUCCESS in 6m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/18a3e6b6b3af426b91cd51b0f6d9e975 : SUCCESS in 10m 05s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c9efe4f32f4e4e2dbdfe8caec564f2e7 : SUCCESS in 5m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c1d5f9ea42144cb79fd9ba4514e32807 : SUCCESS in 7m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f5943fa144c9475295e9582497c14da4 : RETRY_LIMIT in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3d5e6589aa1bba2b4deb7c4f249490d351a4f4c2","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-23 02:37:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1c8eec6355aae335433199320afefb606381abc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-09-24 07:24:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e4ebc8db51c7df2feadf16b8902c03c2f7c0347","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-09-24 07:27:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ebade37a15c704dc3b4cbd24622a49e010853aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 07:49:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c28a452527d04d46bb8d919d17588a37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d80f87a2d84a27b57d86ba977ec5e4 : SUCCESS in 5m 59s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/81802fc0d37549d5b0eac8b9b6334730 : SUCCESS in 8m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7682505e0eb0415b870f659513ad0de8 : SUCCESS in 3m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3cb5f6b0a62947e79398b28fc147a007 : SUCCESS in 4m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/008a7924f9d445d9989063ada2297027 : RETRY_LIMIT in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"317066185de465d28123a1806e963a55ed3237be","author":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"date":"2025-09-24 12:23:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef7eb71fd080dc074b58bd4641f4bbd1057d7d1d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 02:02:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21f1f36d4aabf22a1299eb37e32ae30291da4065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 03:46:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b2f372049d4449093a8c09f2e687622\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08867881722e4a02b5fe466b160b5745 : SUCCESS in 6m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a4974e483e1a4831bd3cef320672735a : SUCCESS in 8m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d361706ca22645d2826fcc1bafc04c60 : SUCCESS in 9m 51s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/770977c829d448688cca0d8277a83696 : SUCCESS in 8m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6fc32a63c39e4ab687d143d09986103b : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a3e0dffb28337bf769f85dd6a0e2848b2a8bb897","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 03:55:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5e7fff2c13afce12e7078349420220e0dc9b61fc","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 03:55:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca966247904b56e0439c7309cb08794f0f75639c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 03:56:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81c75032e26e401f74098bc4e35c5ef17a469ed3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 05:44:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/341dc76ae50a45cc920d603a68786f23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2aac71c913434fb32bb76293c622c2 : SUCCESS in 3m 40s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6d06c9b363a542cca9e88df0a980b4fd : SUCCESS in 11m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5aaf44d990724db3bb64c1c3386fcbfd : SUCCESS in 7m 04s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6c18f084d50a44fcac3fc2749b66fb93 : SUCCESS in 7m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/dba17695871c48f7952b8ea86bc6c3e7 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d43da0758a4afd3be4f4c67797f973720e077ec8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 05:44:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cab012c712c54506cfb9e6b32ee1c77f8c285f69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 05:45:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/270c85ea0dce45248029f7bb9d5a3eac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f0df58fd6f442d2ad13067502c6a63d : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a91a6a857985f9b827eb8cec49c2eabce6a445e","revisions":{"02f12f1a5272638d5a6ffcba825e3ae49785616c":{"kind":"REWORK","_number":1,"created":"2025-09-16 10:07:29.000000000","uploader":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"ref":"refs/changes/59/961359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/961359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/961359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/961359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/961359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/961359/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-09-16 09:53:10.000000000","tz":330},"committer":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-09-16 10:07:00.000000000","tz":330},"subject":"feat: Add Unrescue feature to nova instance actions","message":"feat: Add Unrescue feature to nova instance actions\n\nChange-Id: I64be95aa88fd1e4ed063bbbaef86b3696862b50a\nSigned-off-by: Reet Srivastava \u003creet.srivastava@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/02f12f1a5272638d5a6ffcba825e3ae49785616c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/02f12f1a5272638d5a6ffcba825e3ae49785616c"}]},"branch":"refs/heads/master"},"8a91a6a857985f9b827eb8cec49c2eabce6a445e":{"kind":"REWORK","_number":2,"created":"2025-09-24 07:24:56.000000000","uploader":{"_account_id":38294,"name":"Reet Srivastava","display_name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","username":"reetsrivastava"},"ref":"refs/changes/59/961359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/961359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/961359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/961359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/961359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/961359/2"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-09-16 09:53:10.000000000","tz":330},"committer":{"name":"Reet Srivastava","email":"reet.srivastava@rackspace.com","date":"2025-09-24 07:24:12.000000000","tz":330},"subject":"feat: Add Unrescue feature to nova instance actions","message":"feat: Add Unrescue feature to nova instance actions\n\nChange-Id: I64be95aa88fd1e4ed063bbbaef86b3696862b50a\nSigned-off-by: Reet Srivastava \u003creet.srivastava@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8a91a6a857985f9b827eb8cec49c2eabce6a445e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8a91a6a857985f9b827eb8cec49c2eabce6a445e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
