)]}'
{"id":"openstack%2Fskyline-console~961479","triplet_id":"openstack%2Fskyline-console~master~Iba8f54fc8282e65145409d0c074567b18464bb8c","project":"openstack/skyline-console","branch":"master","topic":"NetQuickStart","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-13 14:35:48.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2026-02-15 23:53:28.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-13 09:12:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Iba8f54fc8282e65145409d0c074567b18464bb8c","subject":"feat: add Network Quick Start functionality","status":"MERGED","created":"2025-09-17 13:13:33.000000000","updated":"2026-02-15 23:54:16.000000000","submitted":"2026-02-15 23:53:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961479-NetQuickStart","meta_rev_id":"7d85cfeb9d223387f043cf0486c0c17ad161fb80","_number":961479,"virtual_id_number":961479,"owner":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-15 23:53:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-14 02:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-14 02:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-17 14:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 02:35:00.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-12-14 21:33:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-12-31 14:12:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"}],"messages":[{"id":"eccb72c22afa15c248928b9cf1ab74019039a87e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-17 13:13:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df245222f0f9ef5f2f849e5ee3aa37e83e6a0549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-09-17 13:17:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd8771a523cacf45d9e0809bf349da4098d3ac3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 14:31:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a15c6b10ae648d0b31dda91dca59e11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52fc4471b9246aaa04b8a9630ae51de : SUCCESS in 6m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/99a0513f0dd342a88d192747c28e3208 : SUCCESS in 11m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7eb1de6fc87e4b30a53600afb749f91e : SUCCESS in 4m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d55541fa0238458da7a2c7d3dae6dc2a : SUCCESS in 10m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/452c0d9cddbc4bafbf1a3c81085abbca : RETRY_LIMIT in 8m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4f2bd4be49624682b8b30f2c54ab1d1b : RETRY_LIMIT in 6m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a42a361ee1fb4bb4b72d5e4e3f6f8f51 : RETRY_LIMIT in 7m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7ba31d27986040eebe05aa22bf19b011 : RETRY_LIMIT in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9cbe920aadb4e4ddb930089d1c40fcfe7ce223df","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-23 02:35:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19043f4635a92599088442397341d52b43771d62","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 21:33:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe0c252b43e7afc2373b58b2096dcd97db72d763","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-12-15 14:22:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14bd374516cc3883b7c8b48b8522527e9065ffe3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-15 14:25:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"842cfe823a354d62fba8f1aab8e6e61a1cbb164f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-02 07:39:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"55825283c460faecdec834d6799e50e2c5f177fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-02 07:43:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"eb558e73e4e0625f59a30cd8794eb3748c5ea21c","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-02 07:45:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e36fbf648d6608ef952e142cd0331cf32c61071e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 09:41:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdb5723aa9594283a45d7692e8555c1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d48cf57ddff3459abb5c05d21ee67df7 : SUCCESS in 6m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7cc1c311153144379d9bf0641fa5fdb9 : SUCCESS in 9m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ff35ac0242bf49f78dd3de685a1209f3 : SUCCESS in 3m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/551026c0025e451b8114539d0146e4af : SUCCESS in 7m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bd4da39153764a9cb9dff7adc9423a13 : FAILURE in 1h 52m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/84f62765346642d685222cbc1489e691 : FAILURE in 1h 25m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/23b0d47250eb462aacf206d86a689251 : FAILURE in 55m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1349e0294c9e40a6b123c47312f60ead : FAILURE in 52m 23s","accounts_in_message":[],"_revision_number":4},{"id":"d108600ae3e49b6887ad38ac9cb52707dab9ef7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-02 10:06:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"611c3a4bc39006feee704b145337e457590675d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 12:05:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2557b9b7c4784841aef31c46d4571503\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab39dfca6ce24331a2adb1f63d0056da : SUCCESS in 4m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/43ee168c7a0d4c55bcc24ccbcb0c4c46 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3596ca85bca44ebc9dff9104696a14ae : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d4cf1376efb044a88085c2d17127fce4 : SUCCESS in 7m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/187149d18b7849c29f2b9d941a00c45e : FAILURE in 1h 53m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a5a824e703b14f398a553dc53ca79d6f : FAILURE in 1h 30m 22s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/14541f7bc0b642868d7f75e708f265a6 : FAILURE in 34m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a29faa6ca25344e5b128559588a57fe3 : FAILURE in 38m 22s","accounts_in_message":[],"_revision_number":5},{"id":"0084e1e613b3fd6d3e9ce2b4678228cf2d45d582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-05 11:21:10.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0e3006798b44d519c673007c75bf9763c0734c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 13:19:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc0d6b53dfb14d5bac8053807c2c9e8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94703f642e748659586d2e5bd8608df : FAILURE in 3m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f33e7380c86248829419294ce4a25945 : SUCCESS in 6m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/53ff8e2abf284d0d9ecc09b9721a3f92 : SUCCESS in 4m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/151db56d294e4065b4b32c671265ed1f : SUCCESS in 4m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b4be756dd434486b8c3c89b1dbfef2ed : FAILURE in 1h 52m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1ead575df5ee496ca18ccb93f779e006 : FAILURE in 1h 01m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3f635d76051b4149a7baaf3a64b671cc : POST_FAILURE in 1h 08m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a8180e8290a5456f8f77030153f38877 : FAILURE in 49m 39s","accounts_in_message":[],"_revision_number":6},{"id":"ffb8ab6f27166009b87c01021e167ad59c4c8c3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-05 13:22:04.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c24c851ac3641f692ab5c7d2bd40f504c9e72229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 15:02:32.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c9bc7fd204d45ce9d694974b23382a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c5fa7e72e734386a41a020697f6b4ac : SUCCESS in 3m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e1f5868504a240778566e7be2ccd4b5f : SUCCESS in 4m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/698670edc0e94c45bc931614dca29031 : SUCCESS in 3m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/89d9d88c9f39441a9cf705bf2ff7983f : SUCCESS in 3m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/93b2609130b642f697e0fbe4cfb895d1 : FAILURE in 52m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/66f6ed826a4b4eb3ac47199cbe6a04df : FAILURE in 1h 28m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bd3db1881dd344c0b3f37cf325fc8f65 : FAILURE in 46m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2c13a5b9ebfb4f0d86f950a96b67f026 : FAILURE in 1h 02m 41s","accounts_in_message":[],"_revision_number":7},{"id":"efec4936d47d32e93e9df6b68a779165753bee61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 06:42:04.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_34517\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"}],"_revision_number":8},{"id":"a283dce3023a2ba7542610cc206b2dcdc3443f8f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 06:47:55.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"125d9a7e0af0f75c7c4bec5d43e12efbe721332c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 08:40:52.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19f11693748748c9bfc48f579138a1ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8e4bd5f6e44121a44889605536a57f : SUCCESS in 5m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/79537900b1934004870121ccc29844fd : SUCCESS in 6m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/095ae0e0f3ac4c62b433b3eccb4d64fa : SUCCESS in 4m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b291672a99674c7dbdc3fdee7b1c2747 : SUCCESS in 3m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/81420724f8634656aca85c42db863062 : FAILURE in 1h 53m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f7a0a35aaa6141388c3a695dc419799b : FAILURE in 48m 35s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a8d38fc5e0b64512b2b07d2f09fd9d04 : FAILURE in 36m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d30efb2c3bea4bc2bb8ec964f6da84f9 : FAILURE in 32m 33s","accounts_in_message":[],"_revision_number":8},{"id":"cc9b646d4991e1f8e2c7d57cab77707623c95937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-22 04:44:44.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"592688683cd2115788ae0128c9627de3d9a2511b","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-22 04:45:24.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bbbd0a1c3fed321f6e03bca1d49e18f5b90d9e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"real_author":{"_account_id":33885,"name":"jessica ridgley","email":"code3jessica@gmail.com"},"date":"2026-01-22 04:46:50.000000000","message":"Patch Set 10: Patch Set 9 was rebased on behalf of \u003cGERRIT_ACCOUNT_34517\u003e","accounts_in_message":[{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"}],"_revision_number":10},{"id":"b7f88bb943696e0c2100c287e14fe440af8b99f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 06:45:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c911d6d25f14981bfa648a0a4471c5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38725ad7793343e08145f7f108de1cbc : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6c810980d3354fe9acd26ee1cc35d862 : SUCCESS in 4m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/cdca6b76acc74594aac1d3bb11dbecef : SUCCESS in 7m 35s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b2213cc75624497ebf5979d79ae5bebe : SUCCESS in 7m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5cdafdce1b9a44c98da82ede6a9cc752 : FAILURE in 1h 55m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/25fcba46355048efbdd7e6a8403cf517 : FAILURE in 1h 28m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2ee38f4f873f4eb6970c542996f92dae : FAILURE in 36m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a2b57dd3a4d44a719629a1dd924251f5 : FAILURE in 39m 00s","accounts_in_message":[],"_revision_number":10},{"id":"d655dafe4923b11e11f7010dca5cde5ac8bd976e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 18:53:49.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_34517\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"}],"_revision_number":11},{"id":"a6025ec8e52d190f191f5f545b867e3163eed2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 20:45:49.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80cbfc824b294c42bf511435091ff9f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139153b0a2214dcfbee0698b12e7ddaf : SUCCESS in 9m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b37ebc02b4d14c0683283f53901f54a5 : SUCCESS in 5m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c391760ce6b5487b847cf921ae2d35a1 : SUCCESS in 7m 57s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d6912871cb3f40809688e5d691f91f35 : SUCCESS in 3m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3cfb26b5d0f2404abab1c6dabc4563ea : FAILURE in 1h 48m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3d06252b3aba4edc9bbbed192503ae6c : FAILURE in 1h 28m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/14946d3529644bbbafd7289e03c4ead8 : FAILURE in 1h 00m 53s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8510a478b9594b07807b472e80e255a8 : FAILURE in 15m 58s","accounts_in_message":[],"_revision_number":11},{"id":"61afb04c57ac08b469f75f5fd49db17ecbf0ad4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-28 13:58:52.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"3f998da6e49edecb2dce75edd299051b7686de06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 16:04:32.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22336ee98d8341ba8ffcba575276aaee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb3017a586943e6a3f2b76bdea88b50 : SUCCESS in 5m 39s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/af9282665b7842fe8cadbcf50a2527d3 : SUCCESS in 10m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/53ea07f373724f08beb775474c9f5085 : SUCCESS in 9m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8ccf3818b61f45d499b269a56fef4743 : SUCCESS in 7m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e33ef49fd43946ce9e56491965fcc386 : FAILURE in 1h 51m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1cf785ccf9a642ed953ebab46bcd67f4 : FAILURE in 1h 24m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/eb0ff3dd518e40168ea03f07089c1173 : FAILURE in 1h 11m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/90a003ae59bc4840954a336b506d0490 : FAILURE in 53m 40s","accounts_in_message":[],"_revision_number":12},{"id":"d370e1cbd7c5dca5bec2db3d1cc275b6de8d0837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-02-13 05:55:38.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"c5a4e117850ebc2c3045358a80c37ef734ca2104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 07:33:02.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78b7aeb2e5144ef984060962f562af69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6493385d664b3298dd483024440a6c : SUCCESS in 3m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d5128f59065347cf8e9285238643fa15 : SUCCESS in 10m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a0b0ebf25d09441c90735e90affd9ca4 : SUCCESS in 7m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f6dede2608194d75bfb8e33523dbbd96 : SUCCESS in 3m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1ec30b80ddb14207987a5ce23afb39d1 : FAILURE in 1h 06m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/785fd01d9f0a496ca2834fe64cfb489f : FAILURE in 1h 30m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/894064de762e496e8de467cceebc3225 : FAILURE in 45m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a959a3e0143048bf8fb8d31b53b53ce1 : FAILURE in 48m 41s","accounts_in_message":[],"_revision_number":13},{"id":"371be3bea022716421f2b827cd8476e59662faf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-02-13 07:59:29.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"869ab5aebbd215493454b4bed0ddf3c154f2340b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 09:08:50.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ddbd5ceb2e64600bd0b781c16da36c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa1f115ed5e45debb5e417f727a2add : SUCCESS in 7m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/03a48637b6f446888cbacf45ed2786ec : SUCCESS in 13m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/087be7d7c88c4d2fa080a4d6d3dc4309 : SUCCESS in 3m 59s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/400654dd7c6c40a78ea377113f51f4c3 : SUCCESS in 4m 58s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fdcf728209b14c99b39cf39f7d148e33 : SUCCESS in 1h 01m 12s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1f0f507567d04de1aaa85d0098971065 : SUCCESS in 47m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/10d0070f8fee4632856df9a7eadcc276 : FAILURE in 44m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6f6eda57402b442e8f3b134d1ae4b67c : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":14},{"id":"a65b0ffd11541d4a4315a1bfe9d840a966e7e201","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:12:18.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"128975e5b207a00a145e911410eeb954c8d45bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-02-13 09:21:12.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"e61afa17dee9dcbbba03743a6ed5978c04d77f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:37:10.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09199d571bbb4be5911085e998342fab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/855088d90f634cb8a7931475cb314caa : SUCCESS in 4m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/371757dd868e4e488b90132d6d2e54dd : SUCCESS in 12m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/73765f9364804b0491559a13b2577f24 : SUCCESS in 8m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9402ea79560e465abccea8e687faa90f : SUCCESS in 9m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c23deb6e9be7407d9705f2feada4bfc4 : FAILURE in 53m 05s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2db582193fe348fdaf49bf51e36fa710 : SUCCESS in 36m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/677607bb33ab43d19c4fdfe0810c0064 : SUCCESS in 54m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b5decfb71ebe41bf86ccc35994de7427 : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":15},{"id":"5a1bc094bafb569ddafef222dc13ab8ecc76434a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-02-13 11:34:24.000000000","message":"Uploaded patch set 16: New patch set was added with same tree, parent tree, and commit message as Patch Set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"e81ddecb53a099e85583ab48df088e0199ec11c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 13:10:38.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc4c57b1ea04e4fb42531cfb2cd19f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa04940907c46dfbe3491383ca1643f : SUCCESS in 4m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fe14ccb9747a4fe7966f81ce8c435a92 : SUCCESS in 11m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6550b51ffb964735a36952c311d10e62 : SUCCESS in 11m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/72a92b9b5b66410ca37f5d0138b938f3 : SUCCESS in 4m 25s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8b76668bdb744acbb773260f0070e4cb : SUCCESS in 1h 30m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b542ffddaba7413dafa99b0793b8b7c9 : SUCCESS in 44m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e68b0af58111410c88d68b0d7fa37170 : SUCCESS in 54m 05s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bb1d4cf1c27b427bbdf60d005dbf6279 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":16},{"id":"bc06d6f619ebcbc5b5037c3354a84847ba836dbb","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-13 14:35:48.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"9ff4c35c4ff4ea2668c8432223964b4bf440f860","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 14:37:55.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4c50d42f826a3ab6a43750e54aa717ba8788b37a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-14 02:14:28.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b824689e798f1168d98fce511ee61f7e716f98f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 02:15:06.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"3f625682f6c2c8a9f043e604268144ce5aaa2e79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 03:36:28.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5be7562bc70040fbbbd65e23acdff6a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b9e7c6aef649a5b2b51a9c06228afd : SUCCESS in 6m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c6403dd3ef794214b039593057033227 : SUCCESS in 11m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/df27a03818b640b8908a14d1ec8e9b04 : SUCCESS in 4m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f6b9de6b0b394f6082bd392aeb23a20c : SUCCESS in 6m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b3a5efa24da544ceb9d2b8c7a8ca0e3d : SUCCESS in 51m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3c0b6078191440d7a74786c6070740c1 : SUCCESS in 1h 00m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3f1f6357e8f546b2b7623934cbfa9ec9 : FAILURE in 43m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/945cd667d0004ada9ac5703885b088fc : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":16},{"id":"b4fe1329a76434b1beedbf49dd96cea905d50c58","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-14 07:18:48.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"0677ed4382f63d660e5da6fbb414bcdb5bcc8bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 08:34:29.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/707b0280615b4fc69c4537e198982038\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0253b36915394cf6b7cf1cd7d84c16f0 : SUCCESS in 6m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1b052021f8be43a6af5eee6e11cd40f5 : SUCCESS in 12m 24s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/448123bf6ab24672904c67c300bb6ea0 : SUCCESS in 7m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/494f81607d374548aa8a0a9f6675fefe : SUCCESS in 5m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b256f7b97b744088abab309001fea6b2 : SUCCESS in 1h 01m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7695b6f8a4ad43c9a55268315303b781 : SUCCESS in 46m 22s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/542c6a7bd7544a658c4003b7b1a03177 : FAILURE in 1h 10m 18s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/98e9ac3e44a843b292257e11c4255dec : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":16},{"id":"2090b2531479260a987c0dbe22c7b65d9d22ed8f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-15 15:34:59.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1bae9e9f82c479664b8d215a2ca08af2caaadccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 17:08:37.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32fdcf8b25024badb1499aea9e0e351b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d522414cc06b47198b223abd7afb55fd : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c53e0490d35b4faab5e9eaf47b45f396 : SUCCESS in 11m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b71d2a3d417f422fb97345b0bc9f66ee : SUCCESS in 8m 05s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3b9a794b087f476e9c28e5d171b9a033 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cbbc01ab29a84e9782d9b73773aa910d : SUCCESS in 1h 27m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1a7b59be66bf4579b3204d4be74bed65 : SUCCESS in 41m 35s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/32efb2cb6d5846778907e193c9fd0a67 : SUCCESS in 52m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ddf01b93490d480784d23a07f84b9692 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":16},{"id":"551f1052e491bff6303137f1c8f0a7fe33613bfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 17:09:05.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"9d668b5aa69ec562e30493cee1713dc22ba0da62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 18:15:25.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e78f65f57274349ac145d02ce9d3455\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057b5b99a49b4137ada23801ba498c29 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ebe0475988f245e4afa4c7faadebb724 : SUCCESS in 4m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d1f4a177486446289ec6c996046605f4 : SUCCESS in 4m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b045c1d07eb847b3b3a1a8176fd7ec4b : SUCCESS in 8m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5508cb63f79148748fd7107a07fce4f7 : SUCCESS in 57m 51s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/14735c6944eb47e6b51b3edfa0ddcc01 : SUCCESS in 59m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d9e80619720c4382babb472b14dccb75 : FAILURE in 52m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/55af1cc694b2411297834213f1275c74 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":16},{"id":"fb83b648016d6c25b1368b4a97df168c4d1de89d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-15 21:21:14.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"98ee3471a267bdbf18b5eda091541791e8332801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 22:49:24.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fc2bdce50774ead8176c3fe71c12873\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4e3ac65e484cc88effeccfc6dae010 : SUCCESS in 5m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bdcd2ab267b3453e92785f1a1d039df7 : SUCCESS in 11m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/de8d0db1be0443128d0f4a88f8a893c6 : SUCCESS in 7m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/edc4eb187b224507bf4a09cdf88bc254 : SUCCESS in 5m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c171c51aa6654a0a8a77a7cb420d44cd : SUCCESS in 1h 21m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/78e2cedbd12b4a4fa83b3edd05bf257f : SUCCESS in 43m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/925803e0ba62437387559bc27702ca99 : SUCCESS in 53m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/38885487555241259a0302af8e37d963 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":16},{"id":"01d044565660b901fe836c5b9130d8624b9dba68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 22:49:51.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"62bd7865d7a9cd8a3acbf9799f93267e6f1368af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 23:53:28.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6fc96182ad742ebbe5e97fee33ab12b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8271ee1ea4a24406a165d99112ec4770 : SUCCESS in 5m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fe7399fffa4f4c96b265fcf032e86679 : SUCCESS in 14m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/818cd4cb6a3c4cf28ba8c091cb938eba : SUCCESS in 4m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d78c27ac12cc472eaea72038f12cd668 : SUCCESS in 7m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8de382d8e170452bb323f32cddacc1fa : SUCCESS in 1h 01m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a9a97970140a49fcb63e56ab613666f2 : SUCCESS in 59m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2f7ddea0cd994e96bbabed254da4a03c : SUCCESS in 37m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/53ea01948df243a1a98f959383b9678c : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":16},{"id":"9171f4223ea347e2a92b8ba7f65c06d1cb65adbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 23:53:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"7d85cfeb9d223387f043cf0486c0c17ad161fb80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 23:54:16.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91ea11b421ec4a31b757106a5d8cc803\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fff2001d1142a3b2cd0b2978418f0c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"bb301b56b10925a7324f765ebf9b0491fb5002ab","revisions":{"acc1676edd8b5ba1559c5db8695713afb87f3ae6":{"kind":"REWORK","_number":1,"created":"2025-09-17 13:13:33.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-17 13:13:12.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nAdd the NetworkQuickStart modal on the Overview page to enable rapid creation of networks, subnets, and routers in a single workflow.\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/acc1676edd8b5ba1559c5db8695713afb87f3ae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/acc1676edd8b5ba1559c5db8695713afb87f3ae6"}]},"branch":"refs/heads/master"},"9da6ceb57e8b11b8ac2ebad4d23fa60ebaa0ded5":{"kind":"REWORK","_number":2,"created":"2025-09-17 13:17:02.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/2"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-17 13:16:52.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nAdd the NetworkQuickStart modal on the Overview page to enable rapid creation of networks, subnets, and routers in a single workflow.\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9da6ceb57e8b11b8ac2ebad4d23fa60ebaa0ded5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9da6ceb57e8b11b8ac2ebad4d23fa60ebaa0ded5"}]},"branch":"refs/heads/master"},"fa4444d77d102d555d430452b2776d74142f4224":{"kind":"REWORK","_number":3,"created":"2026-01-02 07:39:01.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/3"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-02 07:38:18.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fa4444d77d102d555d430452b2776d74142f4224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fa4444d77d102d555d430452b2776d74142f4224"}]},"branch":"refs/heads/master"},"b608d8da8df043db8992e4d412257e5c6d92f2c8":{"kind":"REWORK","_number":4,"created":"2026-01-02 07:43:28.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/4"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-02 07:43:18.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b608d8da8df043db8992e4d412257e5c6d92f2c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b608d8da8df043db8992e4d412257e5c6d92f2c8"}]},"branch":"refs/heads/master"},"6463d80ad61e3cc850106130aa54fa20f0daf3e8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-02 10:06:54.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/5"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-02 10:06:41.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6463d80ad61e3cc850106130aa54fa20f0daf3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6463d80ad61e3cc850106130aa54fa20f0daf3e8"}]},"branch":"refs/heads/master"},"7331b716c5bed99368473b67af1ef0345a1f0942":{"kind":"REWORK","_number":6,"created":"2026-01-05 11:21:10.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/6"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-05 11:20:54.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7331b716c5bed99368473b67af1ef0345a1f0942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7331b716c5bed99368473b67af1ef0345a1f0942"}]},"branch":"refs/heads/master"},"58ca87866833eeeed537663e5b4048868c8d6c58":{"kind":"REWORK","_number":7,"created":"2026-01-05 13:22:04.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/7"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-05 13:21:48.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/58ca87866833eeeed537663e5b4048868c8d6c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/58ca87866833eeeed537663e5b4048868c8d6c58"}]},"branch":"refs/heads/master"},"f4a8aa8bc26f39d6b592b51965bddcca354ffa12":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-01-20 06:42:04.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/79/961479/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/8"}}},"commit":{"parents":[{"commit":"d5a45145061c38e87c1b6a677bf5f9f55f66e8df","subject":"Merge \"Add tooltip for External Gateway option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d5a45145061c38e87c1b6a677bf5f9f55f66e8df"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-20 06:42:04.000000000","tz":0},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4a8aa8bc26f39d6b592b51965bddcca354ffa12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4a8aa8bc26f39d6b592b51965bddcca354ffa12"}]},"branch":"refs/heads/master","description":"Rebase"},"1010e849d62fcc378e8ed1f456ddd94860053886":{"kind":"REWORK","_number":9,"created":"2026-01-22 04:44:44.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/9"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-22 04:44:21.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1010e849d62fcc378e8ed1f456ddd94860053886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1010e849d62fcc378e8ed1f456ddd94860053886"}]},"branch":"refs/heads/master"},"84b70bda5e27af8a817c4f9866e4a64208247738":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-01-22 04:46:50.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"real_uploader":{"_account_id":33885,"name":"jessica ridgley","email":"code3jessica@gmail.com"},"ref":"refs/changes/79/961479/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/10"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-22 04:46:50.000000000","tz":0},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/84b70bda5e27af8a817c4f9866e4a64208247738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/84b70bda5e27af8a817c4f9866e4a64208247738"}]},"branch":"refs/heads/master","description":"Rebase"},"221fabd915eb13d058317357e52e7aaa6a324b7b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-01-27 18:53:49.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/79/961479/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/11"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-27 18:53:49.000000000","tz":0},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/221fabd915eb13d058317357e52e7aaa6a324b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/221fabd915eb13d058317357e52e7aaa6a324b7b"}]},"branch":"refs/heads/master","description":"Rebase"},"fdb33015fd9942aa3bb0303d9d527d73a197d8b3":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-01-28 13:58:52.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/12"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-28 13:58:36.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fdb33015fd9942aa3bb0303d9d527d73a197d8b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fdb33015fd9942aa3bb0303d9d527d73a197d8b3"}]},"branch":"refs/heads/master"},"7b13d49f5efbe520090f75ff4e37719343453453":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-02-13 05:55:38.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/13"}}},"commit":{"parents":[{"commit":"93765736f098df11ccd1a97526fed34795f2b270","subject":"Merge \"Hide Flavor column on Instances page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/93765736f098df11ccd1a97526fed34795f2b270"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-02-13 05:55:38.000000000","tz":0},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b13d49f5efbe520090f75ff4e37719343453453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b13d49f5efbe520090f75ff4e37719343453453"}]},"branch":"refs/heads/master","description":"Rebase"},"27ab2a73d4606976c910025f89338dcc710f9bb8":{"kind":"REWORK","_number":14,"created":"2026-02-13 07:59:29.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/14"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-02-13 07:59:06.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/27ab2a73d4606976c910025f89338dcc710f9bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/27ab2a73d4606976c910025f89338dcc710f9bb8"}]},"branch":"refs/heads/master"},"7bc6dbfd042dd2e66c295629414176ea666ee3da":{"kind":"NO_CHANGE","_number":15,"created":"2026-02-13 09:21:12.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/15"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-02-13 09:20:59.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7bc6dbfd042dd2e66c295629414176ea666ee3da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7bc6dbfd042dd2e66c295629414176ea666ee3da"}]},"branch":"refs/heads/master"},"bb301b56b10925a7324f765ebf9b0491fb5002ab":{"kind":"NO_CHANGE","_number":16,"created":"2026-02-13 11:34:24.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/79/961479/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/961479/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/961479/16"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-02 08:49:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-02-13 11:34:11.000000000","tz":330},"subject":"feat: add Network Quick Start functionality","message":"feat: add Network Quick Start functionality\n\nChange-Id: Iba8f54fc8282e65145409d0c074567b18464bb8c\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bb301b56b10925a7324f765ebf9b0491fb5002ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bb301b56b10925a7324f765ebf9b0491fb5002ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
