)]}'
{"id":"openstack%2Fskyline-console~962050","triplet_id":"openstack%2Fskyline-console~master~Id030676f49e7f3aaddfc2a228a01d22c56838a9e","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-09-30 08:24:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Id030676f49e7f3aaddfc2a228a01d22c56838a9e","subject":"feat: Add Loading form item component","status":"MERGED","created":"2025-09-23 05:53:31.000000000","updated":"2025-09-30 11:21:03.000000000","submitted":"2025-09-30 11:20:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962050","meta_rev_id":"fe6c53d982dc95e6f1974a7340d44947ec437022","_number":962050,"virtual_id_number":962050,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-30 11:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-09-28 03:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-09-28 03:55:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-23 06:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0cf65c11ba9c267d2fe33c3e6554eec1462fd91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-23 05:53:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e22c2e1b2064d659337a64a9499a1bbc9f71c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 06:21:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/702399e0e3be445091879b94c4bf43a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd15d2c325b47bfa2408cc649f3fa95 : SUCCESS in 6m 42s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3265f179683f4439b6c6639c141177c4 : SUCCESS in 4m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3b578d19965d45279230446624b37984 : SUCCESS in 7m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9a74933c456443feabf581dac0009cde : SUCCESS in 7m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/00c7b00ea089432294903d0387637986 : RETRY_LIMIT in 7m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7a3e548eb78a46fbbb3b1406df5de035 : RETRY_LIMIT in 5m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/20df898b7e514b23b26e28c8e9f8d09c : RETRY_LIMIT in 5m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5be1edd7b50441be8a1b7cb1af1d9776 : RETRY_LIMIT in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8ba71ab1899f8423000f77e4b4f7e8019d4bd36d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 02:10:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5695becbcc9d1a406265028e13c1f950b8e712ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 03:43:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce8e1e0ea80c413fa85cd51b0d8c81b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b301c81a39f94ce8ada4ae8f6140e3e8 : SUCCESS in 7m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/aba27ddbfcbf4f0991dbd87b3ac11102 : SUCCESS in 7m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/68d5b71af65648478e14014f1b3e5c3a : SUCCESS in 5m 55s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cfe7a3d085094bf1892ea8644b74422d : SUCCESS in 7m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ec26176ef7c24eff8d97140ccab0f37f : SUCCESS in 1h 21m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bf2cf8487587479882b90e1e3c65ec31 : SUCCESS in 43m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7c9eea59380b451181d2283a74a8a0ee : SUCCESS in 28m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8a47218b0be0471eb111e057130e7283 : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2ca82bb4b5f995f41ec832099d47acc70249845d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 03:55:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68da161ef1ce43fa7c57d8f7900297a9d31061bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 03:56:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6f5cc642528c7b91d54e212bd78694147e3a061","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 04:46:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/765031fcab034f4a83b78bb22aa331e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051223a96b164f1bba18b37ba1390608 : SUCCESS in 3m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/af5c869f827a4ede92b57f9403404d2a : SUCCESS in 6m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3dbb2df871da4c57a1b5fb18562a6f50 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3b77420989614a84b41f779fb2ce2bbe : SUCCESS in 3m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/94bd47bb6e184133a097ac5f22b9f98e : FAILURE in 50m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5eb248de0ec04f079188711b5a7f2f79 : SUCCESS in 29m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/24aae5f49924403a87e15ca51a028830 : SUCCESS in 39m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7720a6727b9d43278bb54c1074019a9f : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":1},{"id":"95060f150f78e712674377fcb6bc3543e77c2233","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 07:00:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73b34502a0e2dd52dadb2f91dc9931e25e81eca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 08:26:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0ef6bd73b6f40aba2b0faeb7ddda66f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/179bfe4f44b7475e90225fc6a5eadf17 : SUCCESS in 5m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/93e9ae8fdcb64bb7bb698a49bfd1b93c : SUCCESS in 4m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/18031278b41249f3bf837255f9a0fd11 : SUCCESS in 7m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/09e7b794ff4441aba8cacf27bbd4a9db : SUCCESS in 3m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/424fedbfcb0b4a9a888bbc23d51aede5 : SUCCESS in 1h 21m 04s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d2fa8d6f401e4d35bd76dd8a3a76c1a9 : SUCCESS in 42m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a1644f094a9741a7b6ea6f524df971a4 : SUCCESS in 41m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/979cdbea0c3043fe9265da45b98e7c72 : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b30580c32181d961545fae414ce3f3132d6ef9d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 08:27:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfe1162ccd93f070cac2c9bb59beb87f143686c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 09:24:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64277205c35047abb6fc22dfc1326a27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656fa97917084e34b3b2429bc7b18e11 : SUCCESS in 3m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/16132054d05c47c992e15fbd2ef808ec : SUCCESS in 4m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/802fb2e7ebfd4d99813a81d1ec967f7d : SUCCESS in 7m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6e8b3c811c304e129833162ae0219b8e : SUCCESS in 3m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/78fc4803cd094b659549f66551905f5d : FAILURE in 53m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0b8b55d5ea0c43cebc8e74fdfd130a3c : SUCCESS in 29m 22s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/aba6b11af9fe4743a913d26d0b21fb75 : SUCCESS in 52m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4125dc4371c942ce9ab60f6adb84e074 : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fb88882a3536937a26041b6c365fba7406cc84e1","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-28 09:25:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1edaafefef07a29ef570de480e35a247ad5dfb51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 10:51:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec418906f6084a38aca9b7bd19898aaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649f26813afb4b65bca0fd958d576d4b : SUCCESS in 4m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7d5dc6de936947d18816c1aaf7ed9a74 : SUCCESS in 6m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/89d6d4621ac24f7384532d6b9ae89e4b : SUCCESS in 4m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6a2b971b436b48029ae3f045ab0315a7 : SUCCESS in 7m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c95f52fc99d646bd82fac28ef41c1412 : SUCCESS in 1h 20m 26s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/68b2811e86ae4526b35f324614b656e6 : SUCCESS in 1h 00m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1d9cbc9eeef3401ba945f44b72b002de : SUCCESS in 51m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3d22e59c71374769acb57d0f37854a07 : SUCCESS in 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8b456364c1a0d11e585b50bad0dc549ba928cc60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 10:51:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40a4413b2d9024a645b80aadae97b9825ef4a7f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 11:55:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f8890bc4b1a4e52bb6d7d47bdd75951\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454431f150c3415493ae30aeea95cacf : SUCCESS in 3m 40s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/952f92ab49ef418cab5f9daf08395028 : SUCCESS in 4m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/642bc3e36a074df485fbea5167614fff : SUCCESS in 4m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c73a6e55ecac471bb33cf1e1db569499 : SUCCESS in 3m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6e36e855d1fa4e4fa05cba7737b18bad : FAILURE in 53m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1356affefa184a3288e976fae5f64e1c : SUCCESS in 59m 09s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8231cb0e831743e081f6924e6a5106f2 : SUCCESS in 54m 54s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b676f358f2464fd2aadd9fe0bc6e29c1 : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2e94f16ceca74ee43577d1adb16f218b81b6ef9a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-09-30 08:24:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04944b1564b5ec2f1e0910ba55ea1035bec43f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 09:55:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09c6f2173843410f945822943c431e5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696bcdc5ec2047488f4bad219bb9f603 : SUCCESS in 5m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/81c7a39f1fdd4b53b2499faa6c0c3cb8 : SUCCESS in 6m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1f564808898a437a87cd0a1f0c917614 : SUCCESS in 3m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/43ebd129aa2c410bb607de3dfc99203a : SUCCESS in 7m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2585329a439148afba64eaf7c993eb2a : SUCCESS in 1h 25m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9b661b9ec1564d7e92d5112aa8e3d2b6 : SUCCESS in 48m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f06de02b74924874942c087f36efc200 : SUCCESS in 54m 45s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d02e9e3584ec44ed880489e6e0a783d0 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c79d56011d5dfaf6ac26938acbcd22d414ffe993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 09:55:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b46f575814d825a397b09381bbdc9c80d8f4bdc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 11:20:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e41b072507aa41a897f8e013e5c27936\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb9d5ffc72e74e729d291330be64e2b4 : SUCCESS in 3m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/886a3b593bd447189d98c0c834b18996 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f889282c3d1842bbbdfbfc9ac9279d43 : SUCCESS in 4m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7e69e2d4aa944e0e85676459ee7612aa : SUCCESS in 7m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/574c6447d1744a8a862e7e062285c164 : SUCCESS in 1h 19m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e1c5268a18e343f88a48ab8801078c99 : SUCCESS in 57m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ad8e425bfa044e58a15a571ed4770d64 : SUCCESS in 51m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/06c323eb23654b02affb994daa994867 : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d3b7c0ab1bb8e4566870a8e6895285dcc96ef4eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 11:20:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fe6c53d982dc95e6f1974a7340d44947ec437022","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 11:21:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c17251ef9946dbad8193b95e522daa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfbcbbfaf3304b1a85beab20bc728d87 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1b66b7e33f2bb398bec73d45e03bfc0abcb064f","revisions":{"d1b66b7e33f2bb398bec73d45e03bfc0abcb064f":{"kind":"REWORK","_number":1,"created":"2025-09-23 05:53:31.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/50/962050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/50/962050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/962050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/962050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/962050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/50/962050/1"}}},"commit":{"parents":[{"commit":"12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1","subject":"feat: Update fixed ip filter for instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12766cf8cf29efa611b59c7bb0f4d3ed372bc0c1"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-09-23 03:44:08.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2025-09-23 05:53:14.000000000","tz":0},"subject":"feat: Add Loading form item component","message":"feat: Add Loading form item component\n\nAdd the component to show loading when the form\u0027s data is still being prepared.\n\nChange-Id: Id030676f49e7f3aaddfc2a228a01d22c56838a9e\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d1b66b7e33f2bb398bec73d45e03bfc0abcb064f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d1b66b7e33f2bb398bec73d45e03bfc0abcb064f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
