)]}'
{"id":"openstack%2Fskyline-console~963502","triplet_id":"openstack%2Fskyline-console~master~I77768b389be5fb05af72c4d63559e6aafceb1104","project":"openstack/skyline-console","branch":"master","topic":"instancename","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-10-29 00:36:07.000000000","reason":"removed on reply"},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2025-11-29 13:54:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77768b389be5fb05af72c4d63559e6aafceb1104","subject":"feat: show instance name in member selection table during load balancer creation","status":"MERGED","created":"2025-10-09 09:28:39.000000000","updated":"2025-11-29 13:54:59.000000000","submitted":"2025-11-29 13:54:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963502-instancename","meta_rev_id":"3098aaecea856d6dac1819deb960172992ac5407","_number":963502,"virtual_id_number":963502,"owner":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-29 13:54:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-10-29 00:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-29 11:14:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-10-29 00:36:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-09 11:02:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-11 09:58:16.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"CC"},{"updated":"2025-10-29 00:36:07.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-11-29 11:14:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"e117135eeed0cba9acd65bb8ebab89062a57ffca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-10-09 09:28:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1e97c37ab62381ca7643a7c7056c2c9e32384e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 11:02:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3000107f0ff4d0e8a114bbe82486844\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afcac0c228c443f286f8b6aa49d79548 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6d9443497a494ce0b62f06609d119e9d : SUCCESS in 9m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d6105723d2594c67bee830c4ac0a187d : SUCCESS in 7m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d6bf3ecce4f7463ea16f92b8668dd081 : SUCCESS in 7m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4e4b6bc65e2d4e89a0c2c245d274c423 : SUCCESS in 1h 20m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d19a3e90e3214d68a15ca82608804f18 : SUCCESS in 1h 12m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7f3c2b25a48d48129b9666b27bcf71c3 : SUCCESS in 50m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/23de74651d5f4f5b893bc8bc5735dbb2 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5bf4c4b6a3e76e82ac80500b1a9dc6e0464df382","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-11 09:58:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c47df8ad9e63da0d457e2a99b10a591c17d19bc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-10-13 08:59:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"298cd48fdf894cabd1b95b7d9c92d0843e91e026","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-10-13 09:07:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"682305e832627ec2dfac177fbe145b2aea8993be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 10:09:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6f27d6378684052a235a034c2fdd045\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205f75bccf4545fa94f4ffe8ac8c999b : SUCCESS in 5m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/403269937bb74599bd994d94d7f8f72e : SUCCESS in 5m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/102e7b8e346848b683b0f60d22ccb2df : SUCCESS in 7m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bc4edfe7585440cea248f0f469d04bc3 : SUCCESS in 3m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d488515fe1914b1d8a69da9e8dc4cda4 : FAILURE in 16m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/eb9bd95467c248bd97a047f9d658b062 : SUCCESS in 29m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b22cc029512347ceb3359b1f5e198e4d : FAILURE in 1h 01m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6b77ed3f255d40c0ad308d873d5241a3 : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":2},{"id":"df6e330e1bf8c39fbdbbb8b6b96ef587219a68e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2025-10-13 11:14:32.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"205b7d8e40b0f0885b351e95ca71b4b45003dd18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 12:46:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9849e29e30344d9d957c19c241c57057\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb5c81fde814aa8a35c74a3f2e72f84 : SUCCESS in 3m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d5781441f7004587b12ad50694c300a5 : SUCCESS in 4m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d95f194ed0d24bf2930a0d5bd58e6042 : SUCCESS in 7m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bce76c27aaf64dcebd0a12fa8b98eeac : SUCCESS in 8m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a8806bcec6e34dec9226f86f8afd3baa : SUCCESS in 1h 26m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/67ed9a81e59d4f879f128bcf564e9e6b : SUCCESS in 41m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9209f807dc374aa099ab4b05f34d365a : SUCCESS in 56m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/41daa373b24d494c8d2b422f15f3e368 : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ffb516fe9fce2cc08c7388d9935ad2d2fd984a7c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-29 00:36:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0566fd5da6d5c8b7c6950da9dfcc5b61142eda0a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-29 00:36:12.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"62a1207daa3d8c0ef665b79663d34a2bffd16c7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 00:36:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"885e00af379e8c5c512acfa1ac9169dcf6d8965e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 02:12:36.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a5745587bb247e9999be947ba06ff3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b3c5057b9d4282869223e5504ea466 : SUCCESS in 5m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7e1022c49a5948f6a4eb0caf65830506 : SUCCESS in 10m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5751c5f29f7a4c6387908b02f141200e : SUCCESS in 6m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f825ea6d2ed8422187b7d1511882b95e : SUCCESS in 4m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/be2c6d95b5d0411a904dcc1e45ad793b : FAILURE in 54m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/09c0dc12e1b047bfb47b6a2f87595d11 : SUCCESS in 34m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3b15e10431eb420ba328df93cd14ff2b : SUCCESS in 38m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/21c6cadf50224e69b53ee1c1d01e91d7 : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":3},{"id":"bf67c51c2b870b9dc32593c85394d33996bc5705","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-29 11:14:57.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"623953028f68d37398c592bc3dc9f30a526654af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-29 12:25:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d527427148c4c4d93b9a73d4f821eb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1ab948382c444b58b80800233b174fa : SUCCESS in 5m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fafce77ee7244142bac8d577f640fb62 : SUCCESS in 6m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e89d91dc69834a4f97bbc9b3d4076cd0 : SUCCESS in 7m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/52507327a66e4490a3e8786c475b87c9 : SUCCESS in 7m 36s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/efc862daa7a3410c95dd83df7dabaeff : SUCCESS in 1h 08m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3c1099900fc946b5910232dd9bc789c1 : SUCCESS in 57m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bee4294018644001b17fd3cd83f9e3f5 : SUCCESS in 53m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/245b15a2a5964527943b7fd8417ef8f3 : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":3},{"id":"4639ecd2a0e08808de564ca3375eed5079a72ba5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-29 12:25:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"07783ab09f4e033ff3a4199820e42c08a5063ccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-29 13:54:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16ff94589104d0fa91d5fee1a2972f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef412aaf5cc436bb5766b603e2160ab : SUCCESS in 5m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e5082aa517964eb280de7fff25a9bcc8 : SUCCESS in 6m 27s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f9f5d625383d4013b1a060c64e72d4a8 : SUCCESS in 3m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/61d3ab1dff41470ba64b992956174eb4 : SUCCESS in 4m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1b4a10aded674355a946a5cf7ac6c30b : SUCCESS in 1h 22m 50s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e3a30dd7fc874c2b9346921c299ee621 : SUCCESS in 30m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2f4ccf75b3fa45009e3ad6dac94838bc : SUCCESS in 38m 18s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/de1e33f8bb3a496bb7b2f80073fba740 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":3},{"id":"10159a157c3b3b7ebae56750caa454a5e385b456","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-29 13:54:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3098aaecea856d6dac1819deb960172992ac5407","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-29 13:54:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4831191a5a60438e83fdb271fd4d0be6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65d359e05e4432eb3254c88027558de : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d76f03ebfa8681e0c7d31798edf32202acf02e6b","revisions":{"d4c97d12a03e01d2812403aa6fb7351cb5351404":{"kind":"REWORK","_number":1,"created":"2025-10-09 09:28:39.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/02/963502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/02/963502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/1"}}},"commit":{"parents":[{"commit":"88ae0e2c059c3f2830086e228ed0fbd05713b218","subject":"Merge \"feat:adds barbican panel(named Key-Manager)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/88ae0e2c059c3f2830086e228ed0fbd05713b218"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-10-09 08:54:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-10-09 08:54:59.000000000","tz":330},"subject":"feat: show instance name in member selection table during load balancer creation","message":"feat: show instance name in member selection table during load balancer creation\n\nChange-Id: I77768b389be5fb05af72c4d63559e6aafceb1104\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d4c97d12a03e01d2812403aa6fb7351cb5351404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d4c97d12a03e01d2812403aa6fb7351cb5351404"}]},"branch":"refs/heads/master"},"3bcae324036c74df13c322574bca0650b20fb083":{"kind":"REWORK","_number":2,"created":"2025-10-13 08:59:25.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/02/963502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/02/963502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/2"}}},"commit":{"parents":[{"commit":"88ae0e2c059c3f2830086e228ed0fbd05713b218","subject":"Merge \"feat:adds barbican panel(named Key-Manager)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/88ae0e2c059c3f2830086e228ed0fbd05713b218"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-10-09 08:54:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-10-13 08:58:50.000000000","tz":330},"subject":"feat: show instance name in member selection table during load balancer creation","message":"feat: show instance name in member selection table during load balancer creation\n\nChange-Id: I77768b389be5fb05af72c4d63559e6aafceb1104\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3bcae324036c74df13c322574bca0650b20fb083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3bcae324036c74df13c322574bca0650b20fb083"}]},"branch":"refs/heads/master"},"d76f03ebfa8681e0c7d31798edf32202acf02e6b":{"kind":"NO_CHANGE","_number":3,"created":"2025-10-13 11:14:32.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/02/963502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/02/963502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/02/963502/3"}}},"commit":{"parents":[{"commit":"88ae0e2c059c3f2830086e228ed0fbd05713b218","subject":"Merge \"feat:adds barbican panel(named Key-Manager)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/88ae0e2c059c3f2830086e228ed0fbd05713b218"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-10-09 08:54:59.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-10-13 11:14:19.000000000","tz":330},"subject":"feat: show instance name in member selection table during load balancer creation","message":"feat: show instance name in member selection table during load balancer creation\n\nChange-Id: I77768b389be5fb05af72c4d63559e6aafceb1104\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d76f03ebfa8681e0c7d31798edf32202acf02e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d76f03ebfa8681e0c7d31798edf32202acf02e6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
