)]}'
{"id":"openstack%2Fskyline-console~963932","triplet_id":"openstack%2Fskyline-console~stable%2F2025.2~I5caa7252651ccb9e2176b243c5c42f241754bf1b","project":"openstack/skyline-console","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"36709":{"account":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"last_update":"2025-10-29 01:36:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5caa7252651ccb9e2176b243c5c42f241754bf1b","subject":"Fix project list for federated users","status":"MERGED","created":"2025-10-14 08:28:38.000000000","updated":"2025-10-29 01:37:22.000000000","submitted":"2025-10-29 01:36:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963932","cherry_pick_of_change":963495,"cherry_pick_of_patch_set":1,"meta_rev_id":"8dc193d7e2002368d8bca2c85cdef90ffb174fb5","_number":963932,"virtual_id_number":963932,"owner":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-29 01:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-10-29 00:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2025-10-21 12:00:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-10-29 00:31:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-14 10:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-29 00:31:24.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"08a67319d1839989434f03c4982aaf0698ad202f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"date":"2025-10-14 08:28:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c897bcbeb0b4bcf9ce5b889a05c4ef54e99b6eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 10:41:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7e227ff44ca418bb828f3a1390b2a1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2634a9c15dca4b6b91196f8e99233221 : SUCCESS in 5m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f1c7636d8e1440c1885ff70b473c5b52 : SUCCESS in 12m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/48e52bb4bf7d429fbc99cfbdf40af06e : SUCCESS in 6m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/817f3b832fc641d3b163e2da417d8aca : SUCCESS in 3m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ab9b8e6bfbbe4ab4a501e4a3d3e6076d : SUCCESS in 1h 36m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9f96fa9c6bb147ebb06a21d2397f6cbe : SUCCESS in 57m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cf2d02f8a27c43bd89c5e2eb138b06dd : SUCCESS in 42m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/90e120d456e5466892f1185493a47991 : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a5a9bd7fbc9819215ce774762fcbf8e8417171db","author":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"date":"2025-10-21 12:00:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"438557891700e1dcbbaef1671f8732bad9d6649a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-29 00:31:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4fa7e40f5e7c3366f6fc861a27e157049d855f94","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-10-29 00:31:36.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f7d78f4cd21c5456ff2e31dbdb0465b68ddfb3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 00:32:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adeb9a03bb80963a5262df94251381a3ec2ea5c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 01:36:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa0bef83432442e9c3e86d8ce31dba5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee692d0d29c64aaca022b4f6dbc1cd80 : SUCCESS in 3m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f9b5a0e401234c05bd942a1911ca8a46 : SUCCESS in 10m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c0fea12bd2444882a6c5bea4618f66b3 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/baa66ebee3b34d2ab620e53602927e9a : SUCCESS in 6m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4249011ad41e4b8aa14fd79cba913e66 : SUCCESS in 1h 02m 43s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/922f7df63df54c71a155034605394582 : SUCCESS in 1h 00m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fa6b2af86ada41c180e422977425b3b1 : SUCCESS in 54m 45s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/70bc1b017b1f4606ad302b8f9b606257 : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7811720ca667154bafdddb65cc2d2d806fd69705","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 01:36:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8dc193d7e2002368d8bca2c85cdef90ffb174fb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 01:37:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5682b1cfef746ff93c66d37fdf52ce1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e1d43b6b1c4d01a83c9f6fcfd3e54c : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a25965a66c1fe6db8a7350e9ed858bda71700e67","revisions":{"a25965a66c1fe6db8a7350e9ed858bda71700e67":{"kind":"REWORK","_number":1,"created":"2025-10-14 08:28:38.000000000","uploader":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"ref":"refs/changes/32/963932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/32/963932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/963932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/963932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/963932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/32/963932/1"}}},"commit":{"parents":[{"commit":"9bd1ce57043b29758d7d4df5b5ec42723c06c83b","subject":"feat: Update package for release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9bd1ce57043b29758d7d4df5b5ec42723c06c83b"}]}],"author":{"name":"Uwe Jäger","email":"uwe.jaeger@valiton.com","date":"2025-10-09 08:26:32.000000000","tz":120},"committer":{"name":"Uwe Jäger","email":"uwe.jaeger@valiton.com","date":"2025-10-14 08:28:38.000000000","tz":0},"subject":"Fix project list for federated users","message":"Fix project list for federated users\n\nFor federated users which can have ephemeral group or projects assignments\nthe Keystone API endpoint /v3/users/{user_id}/projects does not return\nthose assignments, so those users can not switch projects in Skyline\nconsole. Instead we should use /v3/auth/projects which returns all\nprojects the currently logged in user has access to.\n\nChange-Id: I5caa7252651ccb9e2176b243c5c42f241754bf1b\nSigned-off-by: Uwe Jäger \u003cuwe.jaeger@valiton.com\u003e\n(cherry picked from commit d2a0805ae2904db1181ed5832a48f8fca1d73291)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a25965a66c1fe6db8a7350e9ed858bda71700e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a25965a66c1fe6db8a7350e9ed858bda71700e67"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
