)]}'
{"id":"openstack%2Fskyline-console~968940","triplet_id":"openstack%2Fskyline-console~master~I846ff9fae212479805acb60350305cc7550dfe33","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-12-04 14:17:54.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-04 11:30:21.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I846ff9fae212479805acb60350305cc7550dfe33","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2025-11-29 12:18:21.000000000","updated":"2025-12-04 14:19:04.000000000","submitted":"2025-12-04 14:17:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968940","meta_rev_id":"e1df2bc8d29158f0a857dc788e52667dd66ebdd1","_number":968940,"virtual_id_number":968940,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-04 14:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2025-12-01 17:38:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-12-04 03:54:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-12-04 03:54:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-29 13:47:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-29 16:04:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2025-12-01 17:38:29.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-12-04 03:54:48.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"3b18c54ae5f7b5af80205fca002f8003dd8da2cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-29 12:18:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99c3790be7be9dbe52852b247884adb79d50ba55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-29 12:18:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8306afaff1a888858aec2c9bdc0d2a45e69f8105","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-29 12:21:55.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"42b360fac0026ac55f2a4d89802288ebf38fb601","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-29 12:21:59.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"e0632a2c5f91185f69fa1cc3efdf3ff6f2442e97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-29 13:47:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49db208f0dbd458eaf9a540eb44d56d8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ccdf4f086b240049543d9926ed661c3 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7261e98a43324d139b4b672a2e4cd1ea : SUCCESS in 6m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/136cfde70e574e67bded4a2cd274299a : SUCCESS in 10m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d9f23abf06c34ef39bd1bed95325048f : SUCCESS in 7m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/064a4e5515144c43bee47890028015ed : SUCCESS in 7m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0f2981a47c2e484c9364c6d7c2b5a4b7 : SUCCESS in 1h 21m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ce970e9fe83048468972fe5c5e47ff98 : SUCCESS in 1h 02m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/43fa67b1e9f741a4bf97833ba57fde27 : SUCCESS in 55m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2720b14fc500472791eafcf51bfaf61e : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":3},{"id":"cfda2b23a0187623dc9ea70d42f3426bb1169ab3","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-12-01 17:38:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"da0d961abba65d162ff8774332f8d68b849faabf","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-04 03:54:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"86aeed829cc3eb4853e03413b8a1a5aa5415ad26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 03:55:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3a7d900fa7fa404daabf389b88beee5afde60b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 05:24:42.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cc6e40471a94b089d231f217ae0fe6b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c917bfbfc58c413f91f15e28f1523ee9 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe1c7c75465847d9b137f74de36d84d5 : SUCCESS in 5m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/026610c4e77142fcb5fb5e1ede34082f : SUCCESS in 6m 27s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/370a69ff364d4bdeace33ff0f55d5412 : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1649fef9d9c14c628a0045d144c68d98 : SUCCESS in 8m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fd032e97235f4ea68d6c4d213b763203 : SUCCESS in 1h 23m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a0353ca838d6484dbbd4997d6b718f75 : SUCCESS in 42m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c0778f0a48aa4e1794919b460ae72e47 : FAILURE in 40m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/04dea0107dd34f5bbe6ed2d6c183fee5 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":3},{"id":"30107ff36762f8a6c4b7122b23837462895ff6ad","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-04 08:35:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eaa99782598195c2a43338ca40106c46a594be22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 09:35:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b72ef1882bac4f85b1a8b45f56e898d1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9160ed5b7e7040cfb811b982ec4f93d8 : SUCCESS in 1m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac2555dcb0a4a8eabb5dc184d29ad53 : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5a119026c92541c18c35366311e3da5d : SUCCESS in 8m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f3c76acdacbf42128dbe6129453175ba : SUCCESS in 4m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8b56a9bf801c4dc09e5000a021b24bc6 : SUCCESS in 7m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/dbfc599c2cc74c18aebe225b3c60ef79 : FAILURE in 50m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9ae5d83ffdec4414918eb44970fc7bcc : SUCCESS in 30m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/530387bbf51246eb9c06867e28b40b7f : SUCCESS in 40m 48s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8865803907294d6ebf39f7c21f326dd5 : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":3},{"id":"1ee109870b57996661c4b2456aa55ed7e9930290","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-04 09:37:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6a0613569562a3867b2a7d3b7f2dd61cffb07b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 11:07:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/784977f449a0467a8f86a22eb3090cdb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fc7a2689a5eb49e0bbe6933d1ef0a7ca : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbade7daa95649078bef4a5920307fd7 : SUCCESS in 5m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/650650fabe3c4d38b3b8ab2ba11e9e47 : SUCCESS in 10m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/848f5079829741e9b44f946dad019016 : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cad56bad36ac44178e5008880b288cea : SUCCESS in 4m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a8aebfcea6b84706a466fcf4cf502d14 : FAILURE in 1h 23m 04s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/39fd60a94f8c408f8bbd8aa0dc4a27da : SUCCESS in 32m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7dc2f714643347eb815b8ea8c56eeedf : SUCCESS in 48m 54s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8ed38307a3cf41d5977e5cd5af629f58 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":3},{"id":"59b32f5450f6af6db755eab16803706b820f34cf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-04 11:30:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2f9a53d2da2edbed4132761cfc66168dd2ec1ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 12:42:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/430ca908b6cc446d99015cd25c48eb4f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ccf77b750a0544ee8a884c86ed55ef19 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb5f3087bb254fd98ed638db5510ffe0 : SUCCESS in 5m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/13e7cb59271c4101af9bb97e66f86710 : SUCCESS in 8m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d60b0df1095748f7851d3d253e2eccd5 : SUCCESS in 3m 46s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6020a00d12fa47f0a8757cb9c97ec4a1 : SUCCESS in 7m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/43a1c6ac7ad84c2584c416c108cac29a : SUCCESS in 1h 07m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9b4e2c9c0b914e84967b1422b6936818 : SUCCESS in 1h 02m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e31ae9e33bf24d69916b86bf2840070b : SUCCESS in 32m 23s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/44fe0efa5f2547f1812f9e69bfc337b8 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":3},{"id":"5ea073e4cb01025e5d14dbc9b50c617455f4a15f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 12:43:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3eda6599c2571871c31dd63615d4e2cec5a8d2c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:17:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51f4f9c140a24fe68389882e84673862\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa517f1b29114ffd865488687e556106 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2ad67eb0f8438d9d65e31695ead5dc : SUCCESS in 5m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/19af3b0b492842c9aa1b04dcd1281cc7 : SUCCESS in 11m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2e0187619f3c46c78c4932309f20ba50 : SUCCESS in 4m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4a8a6b0545cb473684401118fe431b83 : SUCCESS in 5m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5999874bc9364bc988ee20e992f0e2eb : SUCCESS in 1h 28m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b7229896ffcc4fa29f9415b9ec2e34e6 : SUCCESS in 1h 16m 40s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3e7e0e0a55f84129aa28719e2dfafb91 : SUCCESS in 53m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5cdc5697e7094e40a07fa6d3719745d7 : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":3},{"id":"dd871956815ff32a21051b37bd4b31e83d9c60b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:17:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e1df2bc8d29158f0a857dc788e52667dd66ebdd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:19:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04b37e7d4fbd45b6aaf60555a4f32930\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974986399c8b40459ccc9c2bfc22b676 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"73f16d4e8e5cad71d3da00ac4a984765c7847977","revisions":{"807fd51f68b371b16deaea4554be27a1fe3bdd8e":{"kind":"REWORK","_number":1,"created":"2025-11-29 12:18:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/968940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/40/968940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/1"}}},"commit":{"parents":[{"commit":"97c4e33952b5ded87feaedf5fafaa954f1b05add","subject":"feat: Add Tag functionality to Create Project and tags column in Switch Project table - Implemented ability to add tags when creating a new project - Updated Switch Project table to include a tags column for better visibility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/97c4e33952b5ded87feaedf5fafaa954f1b05add"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-29 12:18:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-29 12:18:21.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I846ff9fae212479805acb60350305cc7550dfe33\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/807fd51f68b371b16deaea4554be27a1fe3bdd8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/807fd51f68b371b16deaea4554be27a1fe3bdd8e"}]},"branch":"refs/heads/master"},"3157b319474fc4e8ab4a4dddcbe4beeeb02c0b7a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-29 12:18:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/968940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/40/968940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/2"}}},"commit":{"parents":[{"commit":"66c9a0efdb9e0b75da11fb3db491038fe64c6ae5","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/66c9a0efdb9e0b75da11fb3db491038fe64c6ae5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-29 12:18:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-29 12:18:48.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I846ff9fae212479805acb60350305cc7550dfe33\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3157b319474fc4e8ab4a4dddcbe4beeeb02c0b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3157b319474fc4e8ab4a4dddcbe4beeeb02c0b7a"}]},"branch":"refs/heads/master","description":"Rebase"},"73f16d4e8e5cad71d3da00ac4a984765c7847977":{"kind":"REWORK","_number":3,"created":"2025-11-29 12:21:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/968940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/40/968940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/40/968940/3"}}},"commit":{"parents":[{"commit":"66c9a0efdb9e0b75da11fb3db491038fe64c6ae5","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/66c9a0efdb9e0b75da11fb3db491038fe64c6ae5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-29 12:18:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-29 12:21:55.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I846ff9fae212479805acb60350305cc7550dfe33\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73f16d4e8e5cad71d3da00ac4a984765c7847977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73f16d4e8e5cad71d3da00ac4a984765c7847977"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
