)]}'
{"id":"openstack%2Fskyline-console~969837","triplet_id":"openstack%2Fskyline-console~master~Iba0a0a0359d2ab530ff19ee7b4e65442185120be","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-12-08 12:50:29.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-08 14:17:43.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Iba0a0a0359d2ab530ff19ee7b4e65442185120be","subject":"Drop basepython","status":"MERGED","created":"2025-12-04 18:50:06.000000000","updated":"2025-12-08 16:49:05.000000000","submitted":"2025-12-08 16:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969837","meta_rev_id":"8dd0e8845e867df83eadfc69fe17c7556327af3b","_number":969837,"virtual_id_number":969837,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-08 16:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-12-08 12:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-12-08 12:50:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-04 19:52:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-07 16:35:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"9115eca8f44817ae9493a57db7b9f1c2db494748","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-04 18:50:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f8ee0fb273f405fc383254a894181c174fcf33a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-04 18:51:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"66aab88c42f915c11f408343d46f1b146185484f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-04 18:51:49.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b9650b57708b6538dee8e8c080244ba2f91f3d8f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-04 18:53:14.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"30ef87c4e1aa4b903a39fe59b407de0a1781a4e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 19:52:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b72fb9b08a77498b8f69aca39aa70d16\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f748de1de6d0409fb02fb08dac3eb8ce : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c1002795f04413ae76d20dc2ca5d10 : SUCCESS in 3m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8194b2fc0b784685a997772b3a8a8eca : SUCCESS in 4m 56s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8202a41dd5994b9792aaca93f700db07 : SUCCESS in 7m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5005a438a6604401a191f464a7849ccd : SUCCESS in 3m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cf55fbef9b0b4fa4ac20bf4c9bb5f253 : SUCCESS in 46m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/60a43f610e3f45e5a2909a24d577dec2 : SUCCESS in 57m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/21f6c935339749cca19dd59416e57d9c : SUCCESS in 49m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8b681749fc50488b964be896c85910c8 : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":4},{"id":"afafa48bd42ca33916ca8a095b40218fee090a9a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-07 16:35:52.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"800395dfff56b7e0502f3eb1dd955d1fd852a1e9","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-08 12:50:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"69ba82f994dae113ca9953e7128ded511e5a1cad","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-08 12:50:35.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ee555f570d6d2d82531b566a732970b2fc240e37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 12:51:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"691bde9edd036a38786c5254b09bc556438380b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 14:15:06.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b952a9502c54411ca6868c03df9d701f\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e079bbcf49244fa4967bc6536659ba4e : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ea5e72a867447f840e0fa75a96b980 : SUCCESS in 5m 55s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e9596c89039b4d569fca02afab22d53c : SUCCESS in 6m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/72bbb6882c184dd78158f43232f8ea9a : SUCCESS in 8m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fcd43f14adf54cbe8777d2fa79132732 : SUCCESS in 8m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/95da887c1319440dbf0f857ca273e747 : SUCCESS in 1h 00m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b6d4698d64d3484badfbfc9791d993c4 : SUCCESS in 1h 19m 19s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/29ab1415d02d4b54b7e26621a369d8a1 : FAILURE in 39m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/76d1fb83e45a414fbe340527dd4d3b6a : SUCCESS in 40m 23s","accounts_in_message":[],"_revision_number":4},{"id":"c88dd83acaf68324ba293132f8e02ef2e7fbc521","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-08 14:17:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aac963163695f439a8a372342e345ed25dc42488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 15:43:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721b7a5757de4cd19092903908c276d7\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca1a988468f84114b719896928d16308 : SUCCESS in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73c42aaede340949eb211da5010ec4c : SUCCESS in 5m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0c1a6bcc216645baa8697666fc67a759 : SUCCESS in 9m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bcbc9bb243294263a60df489ee3106fe : SUCCESS in 6m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/adae3c8df6d945c09ee2548b8cf4f0f0 : SUCCESS in 6m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bc7f158ef50b4c64a747f0aa450b0440 : SUCCESS in 1h 20m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/dcefab3f0b1f47c1bfa4e47b44464d71 : SUCCESS in 59m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/13b93969d624431f932c70a681266b8c : SUCCESS in 29m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/afc8b69b62454b9c95a16fc44807b79f : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":4},{"id":"5f918b51bddce499340cd8bd3a04eb6d1f26e8f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 15:44:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2feee16edbdcaa81a6a6460739276c09a12601cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 16:45:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6db2879b43374eb79e4e45812655af29\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8167b78bb5984e94b787ba498b9a72e8 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b16ae497264d9082e9cfb3e4f1ff6d : SUCCESS in 4m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/38cb3bf9e083458a8fe49dbeb807f742 : SUCCESS in 13m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b2194e83181342889d27c36889d567f2 : SUCCESS in 7m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8174717e6cb84bd79468f946bf13ef1b : SUCCESS in 3m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5cd385a793394a409555bf0388fd5414 : SUCCESS in 1h 00m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c03f61bd5e2043c8b484808422d18f62 : SUCCESS in 50m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/823313f74ed24f14b978294de821ce35 : SUCCESS in 29m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f9830206121c49eba3ed0c270b8f4b3e : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":4},{"id":"18931b25193789ebc158cad7ceece0b5f6ef3add","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 16:45:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8dd0e8845e867df83eadfc69fe17c7556327af3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 16:49:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe00efcb7cd242d79f10e3712e6513dc\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36db66740f67471b9f2c5c198501028b : SUCCESS in 1m 06s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f80183950b1146069eb89765a2a7f6ba : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2bfa4fb75f155fab123463a0f47cc5573c2b1832","revisions":{"05c4193806f84e71d6d22c0185ac953226c5a711":{"kind":"REWORK","_number":1,"created":"2025-12-04 18:50:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/37/969837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/37/969837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/1"}}},"commit":{"parents":[{"commit":"137495be9d66943f92f4da26675f022f81fb12eb","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/137495be9d66943f92f4da26675f022f81fb12eb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-04 18:50:06.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-04 18:50:06.000000000","tz":0},"subject":"Drop basepython","message":"Drop basepython\n\nThis was required to support env with both python 2 and 3 are available\nbut is no longer needed now since python 2 was retired long time ago.\n\nChange-Id: Iba0a0a0359d2ab530ff19ee7b4e65442185120be\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/05c4193806f84e71d6d22c0185ac953226c5a711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/05c4193806f84e71d6d22c0185ac953226c5a711"}]},"branch":"refs/heads/master"},"b902016f365174c0c2aa671ea5f3b5d51bd124e4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-04 18:51:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/37/969837/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/37/969837/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/2"}}},"commit":{"parents":[{"commit":"137495be9d66943f92f4da26675f022f81fb12eb","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/137495be9d66943f92f4da26675f022f81fb12eb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-04 18:50:06.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-04 18:51:00.000000000","tz":0},"subject":"Drop basepython","message":"Drop basepython\n\nThis was required to support env with both python 2 and 3 are available\nbut is no longer needed now since python 2 was retired long time ago.\n\nAlso drop the VIRTUAL_ENV environment which is now automatically set\nby tox.\n\nChange-Id: Iba0a0a0359d2ab530ff19ee7b4e65442185120be\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b902016f365174c0c2aa671ea5f3b5d51bd124e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b902016f365174c0c2aa671ea5f3b5d51bd124e4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c129ecaf1fee1dc44742123f9edf31a53088c402":{"kind":"REWORK","_number":3,"created":"2025-12-04 18:51:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/37/969837/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/37/969837/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/3"}}},"commit":{"parents":[{"commit":"137495be9d66943f92f4da26675f022f81fb12eb","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/137495be9d66943f92f4da26675f022f81fb12eb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-04 18:50:06.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-04 18:51:49.000000000","tz":0},"subject":"Drop basepython","message":"Drop basepython\n\nThis was required to support env with both python 2 and 3 are available\nbut is no longer needed now since python 2 was retired long time ago.\n\nAlso drop the VIRTUAL_ENV environment which is now automatically set\nby tox.\n\nChange-Id: Iba0a0a0359d2ab530ff19ee7b4e65442185120be\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c129ecaf1fee1dc44742123f9edf31a53088c402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c129ecaf1fee1dc44742123f9edf31a53088c402"}]},"branch":"refs/heads/master"},"2bfa4fb75f155fab123463a0f47cc5573c2b1832":{"kind":"REWORK","_number":4,"created":"2025-12-04 18:53:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/37/969837/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/37/969837/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/37/969837/4"}}},"commit":{"parents":[{"commit":"137495be9d66943f92f4da26675f022f81fb12eb","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/137495be9d66943f92f4da26675f022f81fb12eb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-04 18:50:06.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-04 18:53:14.000000000","tz":0},"subject":"Drop basepython","message":"Drop basepython\n\nThis was required to support env with both python 2 and 3 are available\nbut is no longer needed now since python 2 was retired long time ago.\n\nAlso drop the VIRTUAL_ENV environment which is now automatically set\nby tox.\n\nChange-Id: Iba0a0a0359d2ab530ff19ee7b4e65442185120be\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2bfa4fb75f155fab123463a0f47cc5573c2b1832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2bfa4fb75f155fab123463a0f47cc5573c2b1832"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
