)]}'
{"id":"openstack%2Fskyline-console~971667","triplet_id":"openstack%2Fskyline-console~master~Iece9f83ba3cc5d44bbd735236fdeacb6bf2b7abe","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-12-24 02:30:29.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-12-24 02:14:04.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-24 02:30:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iece9f83ba3cc5d44bbd735236fdeacb6bf2b7abe","subject":"Update Russian language","status":"MERGED","created":"2025-12-22 22:31:58.000000000","updated":"2025-12-24 02:31:53.000000000","submitted":"2025-12-24 02:30:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971667","meta_rev_id":"db598085a72af8bf0cbe514ac2144c2119ff7373","_number":971667,"virtual_id_number":971667,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-24 02:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-12-24 02:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-12-24 02:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-22 22:50:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2025-12-22 22:51:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-12-22 22:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-24 00:15:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"45feb43052d30593c1470713dcb2e3375de943aa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-22 22:31:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92d66a6f964f482a535a6a063680cf739f611ae5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-22 22:50:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"56e488d42f756caee5adde82217169f38069f463","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-22 22:50:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d83ab5c9d7b13c2a1cdfa3e5887063262f06d2a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-22 22:50:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8fbb50c0135a0c0e5f7271519d3c7aae5a327c3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-22 22:51:39.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d3959533075217894686b23f98488e4fb57f0a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-22 22:57:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef445fd33b904898836200da9b99f3cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77691e5cf1bc4ed0b1925413af967a43 : SUCCESS in 5m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/02809f22c974427cb6692b41dd3d5cf2 : SUCCESS in 4m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/656f6295b3904b2482016a750efe7ef9 : SUCCESS in 3m 05s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/183c808c0f924215bb6bbe7d6871ba16 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"25660145e139cc6ba972cae8eabd7396074f338f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 23:05:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b51097af0cd8261cefc0157ef1ae014f7bd385b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-23 23:51:04.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"86f035fafda32fdb2dbeb1e3b404810e46c6fb01","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-23 23:51:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"972c610d3b31e8d00cf7e4336d8ddb71bfc77591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 00:05:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60fe3dff82d14254a8743eede86f3b59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da8ec4c84334b02b4b7f01b54bde367 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5a82e821dd9942baa50f6924d119db6c : SUCCESS in 4m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/45c69d7432c4463cb6d18c5a2bb906e4 : SUCCESS in 7m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/dea313c924dd42518468aafc1b2f4e80 : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":4},{"id":"86436ec63ba14534a2f678d441130e1e451d3ec5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-24 00:15:56.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"7ea7cf1c617cc626be81c0323422df815e7766da","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-24 02:14:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f8af56f451a041a3e825ec7113a633a16cb7af59","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-24 02:14:10.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d3ee94ae611d3bf38f484ff03f8c2b4ea251a33e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 02:14:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4198386e6876097375760fab54c518f0b1b915b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 02:30:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db1302da627d4e7290046311b6fa2ecb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78d735acc2804c95943a7c9f4681800b : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9a1aa822e3ed4586868d554b49709be6 : SUCCESS in 10m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/88731a1ce90b459798eedd328c583c01 : SUCCESS in 7m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b5365aefdc4b445e9375b75648ab3f22 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4},{"id":"11ad908575324c4dd4710362d54b298506956440","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 02:30:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"db598085a72af8bf0cbe514ac2144c2119ff7373","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 02:31:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d8e8f0268554b468fec2641b865d450\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819469286fbb4cf7b796ddc5f3c3628e : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"63bba3f48bde34da7584a64d00536aa388862d34","revisions":{"7496bc5cab55f45c72289ffccac2b13b8262ce14":{"kind":"REWORK","_number":1,"created":"2025-12-22 22:31:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/67/971667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/971667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/1"}}},"commit":{"parents":[{"commit":"222e24a774e191dbc5618732a95522bf8ce11492","subject":"Merge \"Fix tox4 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/222e24a774e191dbc5618732a95522bf8ce11492"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-22 22:31:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-22 22:31:58.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iece9f83ba3cc5d44bbd735236fdeacb6bf2b7abe\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7496bc5cab55f45c72289ffccac2b13b8262ce14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7496bc5cab55f45c72289ffccac2b13b8262ce14"}]},"branch":"refs/heads/master"},"ef68666206833f1e219d6121c59f475e764cb2b8":{"kind":"REWORK","_number":2,"created":"2025-12-22 22:50:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/67/971667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/971667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/2"}}},"commit":{"parents":[{"commit":"222e24a774e191dbc5618732a95522bf8ce11492","subject":"Merge \"Fix tox4 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/222e24a774e191dbc5618732a95522bf8ce11492"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-22 22:31:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-22 22:50:04.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iece9f83ba3cc5d44bbd735236fdeacb6bf2b7abe\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ef68666206833f1e219d6121c59f475e764cb2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ef68666206833f1e219d6121c59f475e764cb2b8"}]},"branch":"refs/heads/master"},"455bc16fd4a1fc917a2d8a84066d26b788fca7ca":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-22 22:51:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/67/971667/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/971667/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/3"}}},"commit":{"parents":[{"commit":"222e24a774e191dbc5618732a95522bf8ce11492","subject":"Merge \"Fix tox4 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/222e24a774e191dbc5618732a95522bf8ce11492"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-22 22:31:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-22 22:51:39.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nFix mistakes\n\nChange-Id: Iece9f83ba3cc5d44bbd735236fdeacb6bf2b7abe\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/455bc16fd4a1fc917a2d8a84066d26b788fca7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/455bc16fd4a1fc917a2d8a84066d26b788fca7ca"}]},"branch":"refs/heads/master","description":"Edit commit message"},"63bba3f48bde34da7584a64d00536aa388862d34":{"kind":"REWORK","_number":4,"created":"2025-12-23 23:51:04.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/67/971667/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/971667/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/971667/4"}}},"commit":{"parents":[{"commit":"222e24a774e191dbc5618732a95522bf8ce11492","subject":"Merge \"Fix tox4 error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/222e24a774e191dbc5618732a95522bf8ce11492"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-22 22:31:58.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-23 23:49:39.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nFix mistakes\n\nChange-Id: Iece9f83ba3cc5d44bbd735236fdeacb6bf2b7abe\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/63bba3f48bde34da7584a64d00536aa388862d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/63bba3f48bde34da7584a64d00536aa388862d34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
