)]}'
{"id":"openstack%2Fskyline-console~971917","triplet_id":"openstack%2Fskyline-console~master~I8b4d9567c118ae1cd36f537dba712dc90a7aa8d8","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-12-29 20:30:54.000000000","reason":"removed on reply"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-12-30 12:12:01.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-31 14:01:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b4d9567c118ae1cd36f537dba712dc90a7aa8d8","subject":"Quote - left replaced to Available","status":"MERGED","created":"2025-12-29 18:35:37.000000000","updated":"2025-12-31 14:02:12.000000000","submitted":"2025-12-31 14:01:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971917","meta_rev_id":"cc7a85ebfc96e394f9c727ccfefaf6ad85f06b8a","_number":971917,"virtual_id_number":971917,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-31 14:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-12-29 20:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-12-30 12:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-12-30 12:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-29 18:38:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-12-29 20:04:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-29 20:35:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"e71be5c9c6c2c725861623c146400bbd45100133","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-29 18:35:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"472ad1f5d5edcdb06b359fc71c4bfc2eb9ffc617","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-29 18:36:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ad4ecf64b23bf47aba77889884ab3dabad764bc6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-29 18:38:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"77e04b3eebe210c45ece5c61c39b4d635eafb94f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-29 18:38:27.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"a5c65488b4b71d51120622b0b6c3e32eb4d0cad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 20:04:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc086086ca7f4054811110a6ea4a17b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a80e25364ce45d9bbd3b134bf5c3adb : SUCCESS in 6m 00s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/025db6c130bb4b0da75d4fea21e1d544 : SUCCESS in 4m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3f2813c61cef49c7b8ac9fc76ac6a284 : SUCCESS in 7m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ceaa6436b3094ef885277339c5ef6892 : SUCCESS in 7m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fac584be2c474c4a907823e7347d3fbb : SUCCESS in 1h 22m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4acc955a23b54ffeb694aa09642310d7 : SUCCESS in 43m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d94e67234cea400aa90a84be8047b651 : SUCCESS in 39m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9add45cec69c49c5b66368c3eecc9265 : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":3},{"id":"70df1305902f5b2f7487b6d65cfe5415d0d553a3","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-29 20:30:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ac81e6f33e94e93f276075bf22facf808653a9bb","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-30 12:12:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"47bb668b770f1b7785a359611a841bee83318c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 12:12:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"26228cbfd32d17f65861f9e6e3d2577d31ed0710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 13:10:09.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edc84c16fd784c9fbe2db1b7fb81472d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0de720ea5f4135b58e2faf10aa31f0 : SUCCESS in 6m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3705da3687414ab5b61d2883d28e1a67 : SUCCESS in 6m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a4c89b485c6041358a9e9c6b256e4d96 : SUCCESS in 7m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/db73f5c5ab7e4d4191ca41ac40f77c8e : SUCCESS in 7m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2679c03d1f5b44f685590f91183dca25 : SUCCESS in 49m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/63c9fef535b54353921aebb8ce8383f0 : SUCCESS in 32m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/66c5fc79d35f438ca7a5ce5ee5fc98c3 : FAILURE in 44m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7237bfe98da04fe4a17329a1fc7099eb : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cf5a008bbf3e54064b6e1d0d638dfc985e89c329","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-30 14:03:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"192973e6c853290db2ef8c914ca771bffd381fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 15:05:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b746fe48d0444c4a50c1095d3d38d20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2c5af74c984e13bdb265b4c71329ab : SUCCESS in 6m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/92a69b92a41f402e966e9bafca70f99f : SUCCESS in 4m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3caafca91e864b6ab9cc6c78e953bfee : SUCCESS in 7m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/23e3d808f385494eaae2061e3aab41cb : SUCCESS in 5m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5b5dd66357dc4ce18362d8b870c2f960 : SUCCESS in 1h 01m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ea9e20e654204b229b6cb4a154ea4b26 : SUCCESS in 46m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/14b8565bf58e432aa14e3e74cb24e9f4 : SUCCESS in 38m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/115fb7ec952d4d03ba2a57a17682fb31 : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":3},{"id":"a3f098d472c0093e6cfba442e4413e3f88606264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 15:06:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b4e6c844a4ba7b63489f7d4821174563da4db73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 16:06:42.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3682ab7bdafc4e6285bed65014abf0a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8ebc51681446b9b142edb24bda03d4 : SUCCESS in 5m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0dfdeb2d25c04010a795812b24b4e713 : SUCCESS in 10m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/14237bd9ab84488da0e74a41fcf67ad7 : SUCCESS in 4m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0ee6e5cdcb0343d9be8373491352d444 : SUCCESS in 3m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/26bfe85cac5d4de892bb92cca9714f0a : FAILURE in 52m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e7a786e277fd4ce092e8414153ac93c1 : SUCCESS in 32m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b250726def78497caaae9086fe00755e : SUCCESS in 38m 12s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/15beb491824d4f739fd92b04d67ed176 : SUCCESS in 55m 20s","accounts_in_message":[],"_revision_number":3},{"id":"168b17d976100dcac5ff5c7a86a12cca9799c7d7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-30 18:06:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4aa2be4c7f514f76321941ff6801de03942c9ece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 19:06:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/727ca8f3d18342568747a2c4b7879acc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255b59eae40e4931aa7304693d51e9e1 : SUCCESS in 5m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/361db61abe5f415ab3c029a0b0170138 : SUCCESS in 4m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/31a5557f3195443e88f72f7f0de94d0c : SUCCESS in 7m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d22203b262c743a4ae401761ecad8d82 : SUCCESS in 3m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/43443657837b42bdaa9b70efb75a0f40 : FAILURE in 50m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e0147c2b2d474991be627466b2d1a2f9 : SUCCESS in 32m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e756f6658b4f447db069a264e454fcc4 : SUCCESS in 54m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1aef1d62aa334717b84c7d1732abbca5 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":3},{"id":"5552cc0dea28438bba7253de79697978907df062","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-30 19:14:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cc067c901fe01d169413fbb8d3c78f1736eae04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 20:45:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d876a7f0c59c4802aca65fb0120a5383\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea06dccba1404e62abcd5a345204c840 : SUCCESS in 6m 51s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/74b554ca582a41128ea4efc333d9ab66 : SUCCESS in 4m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5fde6be3d8a048138e0b68bc604668a1 : SUCCESS in 6m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a607af8cfa124c6fba5361752443eceb : SUCCESS in 8m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/26e0efb84de94e19b1e632bd43b7632c : SUCCESS in 1h 25m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1998968e7dd14abfaf83192c5f8fd564 : SUCCESS in 58m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0063d07b003c40ebb8a77dfb4eb74c73 : SUCCESS in 49m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c1a817fc1592463088e126cd248cfe97 : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":3},{"id":"3e55137a5bcf9fe6048e9d0fbbeccf6345f5a918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 20:46:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"470a61e91eb1359e2fffd25cb9b383d7cb7191c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 21:59:56.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b10920a52afe4ab482b65643a1aa1a61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba9f7578f39f48c180452dbe89b534a8 : SUCCESS in 5m 27s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5b1579c5d9624f6097183d13150accca : SUCCESS in 4m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ed8b21dd139c4e86bf28bb4e61e845a1 : SUCCESS in 4m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/06a73d585e9740fb9ba0b7da172b0331 : SUCCESS in 4m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/645cf5b9dda647249d8d5bc4e55202d7 : SUCCESS in 1h 05m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0a846f0a51f944d988c6db10dae80877 : SUCCESS in 41m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ba68c264ff2d490a896b71445d9de898 : FAILURE in 1h 05m 12s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/91fbe43646ac4e158576bb2cfe9b983b : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c476614378d8c038727cfcdb13825eb7d676d150","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-31 01:40:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e19c97c4f08a080ac3098547fddf3b1d0f0124ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 02:42:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e93eb2f0dc8140698e74d8535b0c26a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72400de544e44329abb29adb03082d19 : SUCCESS in 6m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/914d87e456d247599feea51176345670 : SUCCESS in 7m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/df26b772e38848c1a029007cdd192c97 : SUCCESS in 3m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3c3b861dc91a41bdb8ab4ef13bc562f0 : SUCCESS in 7m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/abbc4b4b1b824f0b84862dcc99c2c760 : FAILURE in 50m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/19532e2de81946ce9750355cc7a0ee4b : SUCCESS in 1h 00m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bf7ebe70dc994fbeabca8f67e6add3f8 : SUCCESS in 50m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ad539f6c136c42afa1f76da88d0f9696 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":3},{"id":"a4ca424e8bae2bf4dbc50473f620e98adde912fa","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 02:55:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8d80d427b7532cd33d15eb4072e3c66ace9f268e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 04:28:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2eef71714de41588dd3af1d518b96a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc15fade396645fb9c28b20340ce51f6 : SUCCESS in 5m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8c03b86b1e3742b3a91ade6819e798bb : SUCCESS in 11m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/042c14291a3f48af80eb160b5539c135 : SUCCESS in 4m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3816d6867e254897854102a75c2e09db : SUCCESS in 7m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b8760dea720d4d2da955a7baf34903a5 : SUCCESS in 1h 25m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2ff44a877afe465f810837b973c10180 : SUCCESS in 59m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/98277fa6e4824be2ae315a5b363b3fa5 : SUCCESS in 40m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/56899a1263c5475ca4069c8bc35f8652 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":3},{"id":"8118a00e7343dea6a8b5aff5d13e2e675c6b7c45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 04:28:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"42a30eb119c34be942bd0a8b7334e021d0a0ffa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 05:33:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3403e86e7a84b788c3ee5eab2af4cbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16746d0430a4a999dab630a179bf924 : SUCCESS in 3m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a44cc524becf41fcb03c7786be9b9091 : SUCCESS in 4m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c87944543b8740438dadea8d9bef00d8 : SUCCESS in 6m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5d0e5d399c294ff6839e22bef9243d58 : SUCCESS in 3m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ec655e8a7bf3436b99d088a069f3308b : FAILURE in 53m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/904a7a5a8e87421ea9ee1c4fcd36f550 : SUCCESS in 1h 00m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ecdddc7ad3094dc5b6dfb554869448ee : SUCCESS in 49m 19s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ae985bed613c408da100de9184d4eb1f : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":3},{"id":"526a7aa92773cab12176fb0f329ec5a9fc138963","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-31 10:44:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c97dc3be1db053b998fe2809eed3f0a64a885dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 12:14:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d13021ca4144a59b87c22470f5f5ef7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d16aed665a4c10b4d8e426b9d605a9 : SUCCESS in 5m 33s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d1bafd14faf04c6494a378c6e585068e : SUCCESS in 10m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d3fd8042cfeb4cdb847485dc1376fafb : SUCCESS in 3m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3a0bdef629ac4a4fa4fbfcbaf222018e : SUCCESS in 4m 58s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ab97c57f399744c78fe5f82fab27aab9 : SUCCESS in 1h 24m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4006f699916343e28d1e06392fd1dd5d : SUCCESS in 44m 53s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/da6db1475a904e59a8e9dc504ce27b6e : SUCCESS in 51m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/60160cf61f3248908413a9d90d4a7242 : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":3},{"id":"5bd9f5feaf8bc0080db5a16ada124f20b90d82dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 12:14:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0c7559cd78c904684884dbadf7dac625f8580b9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 14:01:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd8f3d2ae584f8abe38b712e80ebf9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7338d358d74788acace7d9843e301d : SUCCESS in 3m 39s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8e2b9229419f48bb9029216bb0147203 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fac243c5767141e095e395fd5cdaf380 : SUCCESS in 3m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/23170dcc79e84a89b0dfe9be0980fef8 : SUCCESS in 7m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9d6efc45d7c24ba2aac5f3f227801516 : SUCCESS in 1h 22m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2f66cdbc186443d6a82d188b95c4629e : SUCCESS in 41m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/405ddfc7f2714c92b31ac4a7e8d00c1c : SUCCESS in 28m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/348d8e33d77e4f80a7a3a9871db0744a : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ab74de25fd88e80b9dafe35f10d2e966b0cd75bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 14:01:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cc7a85ebfc96e394f9c727ccfefaf6ad85f06b8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 14:02:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/621e0fb461dd4d899a0a79ea89c8085d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fa58da4a1c48a89b01c9c56484e6e9 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6d014f8d145311eed3bdf98887ce3c95abade228","revisions":{"21c48e41aafee2970aeda2e1f44bfa62ba078467":{"kind":"REWORK","_number":1,"created":"2025-12-29 18:35:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/17/971917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/17/971917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/1"}}},"commit":{"parents":[{"commit":"9260ac2724cec888eda2e1bad25d98de46d3e044","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9260ac2724cec888eda2e1bad25d98de46d3e044"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-29 18:35:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-29 18:35:37.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I8b4d9567c118ae1cd36f537dba712dc90a7aa8d8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/21c48e41aafee2970aeda2e1f44bfa62ba078467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/21c48e41aafee2970aeda2e1f44bfa62ba078467"}]},"branch":"refs/heads/master"},"5d883cbf58bfcba61aa50d3b141ed23c51c8e1da":{"kind":"REWORK","_number":2,"created":"2025-12-29 18:36:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/17/971917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/17/971917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/2"}}},"commit":{"parents":[{"commit":"9260ac2724cec888eda2e1bad25d98de46d3e044","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9260ac2724cec888eda2e1bad25d98de46d3e044"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-29 18:35:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-29 18:36:54.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I8b4d9567c118ae1cd36f537dba712dc90a7aa8d8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d883cbf58bfcba61aa50d3b141ed23c51c8e1da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d883cbf58bfcba61aa50d3b141ed23c51c8e1da"}]},"branch":"refs/heads/master"},"6d014f8d145311eed3bdf98887ce3c95abade228":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-29 18:38:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/17/971917/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/17/971917/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/17/971917/3"}}},"commit":{"parents":[{"commit":"9260ac2724cec888eda2e1bad25d98de46d3e044","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9260ac2724cec888eda2e1bad25d98de46d3e044"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-29 18:35:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-29 18:38:13.000000000","tz":0},"subject":"Quote - left replaced to Available","message":"Quote - left replaced to Available\n\nI think this will more accurately display information about the remaining resources.\n\nChange-Id: I8b4d9567c118ae1cd36f537dba712dc90a7aa8d8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6d014f8d145311eed3bdf98887ce3c95abade228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6d014f8d145311eed3bdf98887ce3c95abade228"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
