)]}'
{"id":"openstack%2Fskyline-console~971968","triplet_id":"openstack%2Fskyline-console~master~I5b3f2e18974222f88f44394184825b8f0acb81bc","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-04 05:39:01.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-04 04:08:57.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2026-01-04 05:39:01.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-04 05:39:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b3f2e18974222f88f44394184825b8f0acb81bc","subject":"Rename field “Available Zone” to “Availability Zone”","status":"MERGED","created":"2025-12-30 12:15:54.000000000","updated":"2026-01-04 05:39:45.000000000","submitted":"2026-01-04 05:39:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971968","meta_rev_id":"711aa4c84293906586add8b84833a95720b2dcc6","_number":971968,"virtual_id_number":971968,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-04 05:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-04 04:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-31 02:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-04 04:08:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-30 14:13:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-30 14:20:45.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-12-30 14:20:45.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-31 14:23:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"b7f59276b2d359a416f02a14442a8befd50c95fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-30 12:15:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca7a1ed59a33777d7754500e0f6909c6eff24dd5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-30 12:38:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c6254af6a31e621a33bffd88860cad656f9fcbb7","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-30 13:47:02.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"08ff38d801d7964ce57154a62e05aeb2d8534801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 14:13:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f77554fe029428d93b8bd03c3bad665\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5719d31304c49a89966934de69d3a82 : SUCCESS in 3m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e35807636cc949cfba754ac4d8770c71 : SUCCESS in 10m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8a4b8911a4a743809a63773250a7daa5 : SUCCESS in 7m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4f31545065a44197bcd9b06ac6bc5d23 : SUCCESS in 4m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2c164c46e10e423f939292ca1d5636f2 : SUCCESS in 1h 27m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6c196b55cb75490dbd0248d91263d9e0 : SUCCESS in 49m 53s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b911be03770e47f599c8d2f6816ebe3b : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":2},{"id":"df2fd02ac99d52cc5b395d7793db770ddaf45b48","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-30 18:06:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4726c5af9c1fce2852770c928ad43eaea3090053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-30 18:37:09.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"04ff7cd00e2aab3db503b5a3c46da3344a8456dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-30 18:37:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"985a7e92ea492397e062eba4569bf3f24b9b3531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-30 18:37:48.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"49d2e2459d1c49c99f0efb1687a1e921f3ad541a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-30 18:37:55.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"66526a907bd30f2d68edea222df633faa919bee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 19:42:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5af71a2142e4272bede096f0ed2cb7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8741f8f868a64010a50cbfb5691a6493 : SUCCESS in 4m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8c597f1fa60c48218cf298f13fd348b6 : SUCCESS in 6m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/cc5fa3ee43c246e4806c7020547b51a8 : SUCCESS in 8m 15s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/dda0d01092a24a8da16a2566a82c3183 : SUCCESS in 4m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bdb7ce9782d341b79d516292e7350be2 : SUCCESS in 57m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/77abc469479b4e09bb116c5654d52e9b : SUCCESS in 1h 00m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/468218ad6af3430d8c1a9d16fa0bdfd7 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":5},{"id":"7b35d3d6ab560911ee0e23685f1181e7e7d5f339","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-31 01:42:15.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9e7fac63eab204c3ff71bbec1d246bc23854a804","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 02:11:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"728e662395a021209a4850bc44d3f03bb664c084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 02:11:35.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"908e0288b0194a55cd05e40e40e67003ee120e83","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 02:12:15.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"af3b5061071dbfd8a3b6782ac4caa718e9c159ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 03:43:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/033167c47eb14aa0ad35fa8efbd08ca8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bdc854d99774031a9b7368c760e2806 : SUCCESS in 6m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0d274ad9aa2b44908c4ca190fbad51d7 : SUCCESS in 9m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e8136f3dbdd0407a89e1514a071947fb : SUCCESS in 4m 46s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5bce0e8da0334bf1af5c2d8cc9029908 : SUCCESS in 3m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/47cd26bc68d24e18a6b1a53d65ec8b76 : SUCCESS in 1h 02m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ba8707daff9e450a87cc7c76b00c3b5d : SUCCESS in 58m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c70f633a06ec4054836347fa250cc8a3 : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":6},{"id":"967fbe7a4a45029d028340e4d615699d266bcc64","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-04 04:08:57.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7b2f77437631f4a156d674bd097a61f643d001bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 04:09:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"394d78704d9c225974c0034cb6df407bacd86e49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:39:01.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cbe8d64ec6f4479801d928f6b8be1d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86a1ff15ac1a47cdad017eebf3d08fb3 : SUCCESS in 3m 11s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/90ba26f50a64414f9887ae20431499db : SUCCESS in 10m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ab3b353924a4498b91b340a00166de3d : SUCCESS in 3m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ee1cd94ed0d84c9ea16c1d2e3a588535 : SUCCESS in 6m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9b0f7f77fbaf4c43a05dc5b84733c903 : SUCCESS in 1h 20m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ec3645251e09432ebc0b85d580ee751c : SUCCESS in 1h 00m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0c268c263f944679bb4a1ae924b88258 : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":6},{"id":"2a300c249854f6b685603181cfa6475df99994d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:39:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"711aa4c84293906586add8b84833a95720b2dcc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:39:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b06c243221c4489cbcae5ea7a1898716\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3097f36ee91646c59eb799aa6623775e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"080bcee61204204cd7a55617a257ed6ef4945565","revisions":{"70d98fdf0d3e5da1b94c4a8b931fdea904b1f3f3":{"kind":"REWORK","_number":1,"created":"2025-12-30 12:15:54.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/68/971968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/971968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/1"}}},"commit":{"parents":[{"commit":"9260ac2724cec888eda2e1bad25d98de46d3e044","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9260ac2724cec888eda2e1bad25d98de46d3e044"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-30 12:15:54.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-30 12:15:54.000000000","tz":0},"subject":"Rename field “Available Zone” to “Availability Zone”","message":"Rename field “Available Zone” to “Availability Zone”\n\nCorrect a typo in the UI by renaming the field “Available Zone” to “Availability Zone”\nfor terminology consistency.\n\nChange-Id: I5b3f2e18974222f88f44394184825b8f0acb81bc\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/70d98fdf0d3e5da1b94c4a8b931fdea904b1f3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/70d98fdf0d3e5da1b94c4a8b931fdea904b1f3f3"}]},"branch":"refs/heads/master"},"e65c5a72971a6b6cfb38089175ee90711ab334b9":{"kind":"REWORK","_number":2,"created":"2025-12-30 12:38:50.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/68/971968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/971968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/2"}}},"commit":{"parents":[{"commit":"9260ac2724cec888eda2e1bad25d98de46d3e044","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9260ac2724cec888eda2e1bad25d98de46d3e044"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-30 12:15:54.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-30 12:38:40.000000000","tz":0},"subject":"Rename field “Available Zone” to “Availability Zone”","message":"Rename field “Available Zone” to “Availability Zone”\n\nCorrect a typo in the UI by renaming the field “Available Zone” to “Availability Zone”\nfor terminology consistency.\n\nChange-Id: I5b3f2e18974222f88f44394184825b8f0acb81bc\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e65c5a72971a6b6cfb38089175ee90711ab334b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e65c5a72971a6b6cfb38089175ee90711ab334b9"}]},"branch":"refs/heads/master"},"980c70f975454ccaaf3d666b8f70e65e13aab0d7":{"kind":"REWORK","_number":3,"created":"2025-12-30 18:37:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/68/971968/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/971968/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/3"}}},"commit":{"parents":[{"commit":"9260ac2724cec888eda2e1bad25d98de46d3e044","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9260ac2724cec888eda2e1bad25d98de46d3e044"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-30 12:15:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-30 18:37:07.000000000","tz":0},"subject":"Rename field “Available Zone” to “Availability Zone”","message":"Rename field “Available Zone” to “Availability Zone”\n\nCorrect a typo in the UI by renaming the field “Available Zone” to “Availability Zone”\nfor terminology consistency.\n\nChange-Id: I5b3f2e18974222f88f44394184825b8f0acb81bc\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/980c70f975454ccaaf3d666b8f70e65e13aab0d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/980c70f975454ccaaf3d666b8f70e65e13aab0d7"}]},"branch":"refs/heads/master"},"5971de582b58b71a25a8e583f7cfe5690ae8f472":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-30 18:37:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/68/971968/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/971968/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/4"}}},"commit":{"parents":[{"commit":"0f1fac5eaea1ace74861827b978cb8351233b8bc","subject":"Update Russian language","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f1fac5eaea1ace74861827b978cb8351233b8bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-30 12:15:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-30 18:37:14.000000000","tz":0},"subject":"Rename field “Available Zone” to “Availability Zone”","message":"Rename field “Available Zone” to “Availability Zone”\n\nCorrect a typo in the UI by renaming the field “Available Zone” to “Availability Zone”\nfor terminology consistency.\n\nChange-Id: I5b3f2e18974222f88f44394184825b8f0acb81bc\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5971de582b58b71a25a8e583f7cfe5690ae8f472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5971de582b58b71a25a8e583f7cfe5690ae8f472"}]},"branch":"refs/heads/master","description":"Rebase"},"ad46949095ab656e6146e3ae6a1ff5508ed82b6e":{"kind":"REWORK","_number":5,"created":"2025-12-30 18:37:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/68/971968/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/971968/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/5"}}},"commit":{"parents":[{"commit":"0f1fac5eaea1ace74861827b978cb8351233b8bc","subject":"Update Russian language","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f1fac5eaea1ace74861827b978cb8351233b8bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-30 12:15:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-30 18:37:48.000000000","tz":0},"subject":"Rename field “Available Zone” to “Availability Zone”","message":"Rename field “Available Zone” to “Availability Zone”\n\nCorrect a typo in the UI by renaming the field “Available Zone” to “Availability Zone”\nfor terminology consistency.\n\nChange-Id: I5b3f2e18974222f88f44394184825b8f0acb81bc\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ad46949095ab656e6146e3ae6a1ff5508ed82b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ad46949095ab656e6146e3ae6a1ff5508ed82b6e"}]},"branch":"refs/heads/master"},"080bcee61204204cd7a55617a257ed6ef4945565":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-12-31 02:11:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/68/971968/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/971968/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/971968/6"}}},"commit":{"parents":[{"commit":"41db86bb518384b57d7eea2c3693428c8362287c","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/41db86bb518384b57d7eea2c3693428c8362287c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-30 12:15:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 02:11:35.000000000","tz":0},"subject":"Rename field “Available Zone” to “Availability Zone”","message":"Rename field “Available Zone” to “Availability Zone”\n\nCorrect a typo in the UI by renaming the field “Available Zone” to “Availability Zone”\nfor terminology consistency.\n\nChange-Id: I5b3f2e18974222f88f44394184825b8f0acb81bc\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/080bcee61204204cd7a55617a257ed6ef4945565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/080bcee61204204cd7a55617a257ed6ef4945565"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
