)]}'
{"id":"openstack%2Fskyline-console~972027","triplet_id":"openstack%2Fskyline-console~master~I58476b76e0b8a68a4af52ef88dceb27724ccf940","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-04 05:29:23.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-04 04:08:09.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-04 05:29:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I58476b76e0b8a68a4af52ef88dceb27724ccf940","subject":"Port Detail - update size for Base Info block","status":"MERGED","created":"2025-12-31 14:57:29.000000000","updated":"2026-01-04 05:30:58.000000000","submitted":"2026-01-04 05:29:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972027","meta_rev_id":"5635b93f50bef063832e09b734e181db81dc3fa5","_number":972027,"virtual_id_number":972027,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-04 05:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-04 04:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-04 04:08:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-31 15:24:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-12-31 15:24:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-12-31 16:08:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b01a5feff9457d7cc9f2ca596db03ed8428b1b40","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 14:57:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee1b561cf87af39d25ea739cd5141f09eec068fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:06:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b1de73be0ed608b923e9e090e40c7477b2e296bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:09:28.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6b7c256285d360b2a9e43d6536125001daed42cc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:18:22.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c6b6e5637bb93bed6034ace22cf7111d4ab7779f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:24:25.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"474b7f7bb54c93b52923eee84905728daa2d830e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:24:34.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"e661e8b659b62edff1396b885f44aeedf1437b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 16:08:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f5f4120581c44fa9db30025e4017573\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad5cf61c1744d22af23fb4a988228ef : SUCCESS in 5m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3b6fe962f08147d1aaa1fadb2fcd225d : SUCCESS in 10m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1340526db68a47128d4057001bb06d06 : SUCCESS in 7m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2f9834d2acc7479c9c8bc235d3696b9a : SUCCESS in 7m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bd3725c9736f4440b5016cca0d434ad6 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":5},{"id":"6afc6815f2e25c9709d7ad827442b5df16331535","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 21:37:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4ef803c23de538e4f895ba93530cf7775648795a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-04 04:08:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4a99affada16da532683875e12134e2980e06a86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 04:08:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dd41f0521734e1b19611bed370394b38d993314e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:29:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e45929ff4ce34a4ca0b3a299445fff05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bff2246c50b4cf380c8aabaaf25bd81 : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/dd990cdfde7a4fd5ad5d2043787bb00c : SUCCESS in 11m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/eaa33a74f634426fa78a9b1cce89b598 : SUCCESS in 7m 30s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b7312bf50bc24628bc4ad0043225d18b : SUCCESS in 4m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/13cd90ae31fc45c0bca5309fbe91fa9b : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":5},{"id":"1a5d0f6253b7cabfbcd2429efd2fb9db5fc61eaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:29:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"5635b93f50bef063832e09b734e181db81dc3fa5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:30:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ce8075fe1d489d842a29cbdddf3f1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac3e1336feea4141aaaa434b99ff746a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"953ff53c2ff9b68dfdfda36b0398f018143eeda4","revisions":{"51327bb354cac9b69ed4ea38b29cea4581bf84a9":{"kind":"REWORK","_number":1,"created":"2025-12-31 14:57:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/27/972027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/27/972027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/1"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 14:57:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 14:57:29.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I58476b76e0b8a68a4af52ef88dceb27724ccf940\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/51327bb354cac9b69ed4ea38b29cea4581bf84a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/51327bb354cac9b69ed4ea38b29cea4581bf84a9"}]},"branch":"refs/heads/master"},"3f57f4c128907d9eac728157a393b892948b0fb9":{"kind":"REWORK","_number":2,"created":"2025-12-31 15:06:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/27/972027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/27/972027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/2"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 14:57:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:06:18.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I58476b76e0b8a68a4af52ef88dceb27724ccf940\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f57f4c128907d9eac728157a393b892948b0fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f57f4c128907d9eac728157a393b892948b0fb9"}]},"branch":"refs/heads/master"},"19c0eec49d5b224a5a8d4064136dfe6b8362f2e0":{"kind":"REWORK","_number":3,"created":"2025-12-31 15:09:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/27/972027/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/27/972027/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/3"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 14:57:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:09:26.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I58476b76e0b8a68a4af52ef88dceb27724ccf940\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19c0eec49d5b224a5a8d4064136dfe6b8362f2e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19c0eec49d5b224a5a8d4064136dfe6b8362f2e0"}]},"branch":"refs/heads/master"},"e96c993808854c93ca1a4d3fbab6327c6416d931":{"kind":"REWORK","_number":4,"created":"2025-12-31 15:18:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/27/972027/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/27/972027/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/4"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 14:57:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:18:22.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I58476b76e0b8a68a4af52ef88dceb27724ccf940\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e96c993808854c93ca1a4d3fbab6327c6416d931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e96c993808854c93ca1a4d3fbab6327c6416d931"}]},"branch":"refs/heads/master"},"953ff53c2ff9b68dfdfda36b0398f018143eeda4":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-12-31 15:24:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/27/972027/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/27/972027/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/27/972027/5"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 14:57:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:24:25.000000000","tz":0},"subject":"Port Detail - update size for Base Info block","message":"Port Detail - update size for Base Info block\n\nIt is necessary to make the block wider for languages ​​where phrases are long.\n\nScreenshot:\nhttps://anfimov.name/openstack/bug_972027.png\n\nChange-Id: I58476b76e0b8a68a4af52ef88dceb27724ccf940\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/953ff53c2ff9b68dfdfda36b0398f018143eeda4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/953ff53c2ff9b68dfdfda36b0398f018143eeda4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
