)]}'
{"id":"openstack%2Fskyline-console~972028","triplet_id":"openstack%2Fskyline-console~master~I746c13b80dc6fbd6e795dfc712a5fcea80c0decb","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-04 05:29:21.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-04 04:07:24.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-04 05:29:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I746c13b80dc6fbd6e795dfc712a5fcea80c0decb","subject":"Key Manager - hide for admin if barbican not used","status":"MERGED","created":"2025-12-31 15:39:14.000000000","updated":"2026-01-04 05:30:11.000000000","submitted":"2026-01-04 05:29:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972028","meta_rev_id":"1510402c8f4a7d2c8f2ca82e4288b61f69e49816","_number":972028,"virtual_id_number":972028,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-04 05:29:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-04 04:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-04 04:07:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-31 15:43:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-12-31 15:43:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-12-31 17:15:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20e7fedcb4f79f68a994e48855fc979940d26020","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:39:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8151f4d904534a57f379554701ca24b13f530552","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:42:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"654fb44b0efaff92f7d3cd67aa62be81b289d175","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:42:59.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"819a8c0b0a1d8927b287e03e81a6894c2856f600","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:43:04.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"51fc4bb44eac3acbdc289c404af4870f2fe1eb62","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:43:19.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"26b981dfd4412b62df56df591a3986a76868fa0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-31 15:44:18.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"98787bdd93fb67d6cb64c54b7ee7df47c9c7626d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 17:15:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eefc5da7ece45b799854f1ad683c73f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005832e527164bd39fd88fe8e73e687e : SUCCESS in 6m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/225ddacccf104456a2398a033c3ae33d : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/dda35c2f419d4d1caa10bd03dfe64093 : SUCCESS in 8m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/449ab3161db94b1687772edcbb567f58 : SUCCESS in 7m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c97fe712399845f1b40907c7d0c59448 : SUCCESS in 1h 26m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a17a8784524a44bda0735d01fcaa7ee3 : SUCCESS in 1h 01m 13s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0fc4d07c1ddb44919dd6ad7f428523c6 : SUCCESS in 54m 06s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a8e550b97a224a40bbc7b4026c37ad8d : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":5},{"id":"9a75aaa7d4f91785ed2dcc3d5eb812290689a16a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-04 04:07:24.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4ada28c13ceeda97459a646ef2c5653a933bb5e6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-04 04:07:30.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8b21e53583079c04795988af4267a5fecc995fdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 04:08:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b2c7b606e9bb325dd5a193eb118ab5075c163b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:29:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e903438d3096455480fbb8856b23712c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee69742a0524eedaaec14e751fe29c6 : SUCCESS in 5m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/03da904e359845b58d1b59220f898ffd : SUCCESS in 4m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9f3a6b9792b74d5f8714763ab3925ae2 : SUCCESS in 6m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c2664985742f42078b209b9bbe640bab : SUCCESS in 6m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0e8d088c4c7945098b097249893be677 : SUCCESS in 1h 20m 05s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5aa739f37b2e443f8b05424d02a4867e : SUCCESS in 59m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2b88ed0b6c01418d93b5603fa43c2861 : SUCCESS in 37m 19s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f4297115c98b4311938e6aef0e9e1ae2 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":5},{"id":"0f4502647b13aa976bf714dbb030fd1e94620ba6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:29:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"1510402c8f4a7d2c8f2ca82e4288b61f69e49816","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 05:30:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f625ce2f144e3cbde6d4800cac3889\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bf02d8ba93b4cf5959ad9837a27fcfa : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"32aa56dab233d7a42d2324ad56797a8c4f2e6ab9","revisions":{"accd3d9afe80316cde5fd907c0f59c833805a91a":{"kind":"REWORK","_number":1,"created":"2025-12-31 15:39:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/972028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/972028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/1"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:39:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:39:14.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I746c13b80dc6fbd6e795dfc712a5fcea80c0decb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/accd3d9afe80316cde5fd907c0f59c833805a91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/accd3d9afe80316cde5fd907c0f59c833805a91a"}]},"branch":"refs/heads/master"},"7a8cc84f191c72def9e2286e73d333f324e176dd":{"kind":"REWORK","_number":2,"created":"2025-12-31 15:42:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/972028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/972028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/2"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:39:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:42:10.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I746c13b80dc6fbd6e795dfc712a5fcea80c0decb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a8cc84f191c72def9e2286e73d333f324e176dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a8cc84f191c72def9e2286e73d333f324e176dd"}]},"branch":"refs/heads/master"},"4f24e2bbb749e6a458a65ed8cb7861e9a6734653":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-31 15:42:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/972028/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/972028/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/3"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:39:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:42:59.000000000","tz":0},"subject":"Key Manager - hide for admin if barbican not used","message":"Key Manager - hide for admin if barbican not used\n\nChange-Id: I746c13b80dc6fbd6e795dfc712a5fcea80c0decb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4f24e2bbb749e6a458a65ed8cb7861e9a6734653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4f24e2bbb749e6a458a65ed8cb7861e9a6734653"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5b116f92e03c81456035ddcc42eebc78320dc147":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-12-31 15:43:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/972028/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/972028/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/4"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:39:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:43:04.000000000","tz":0},"subject":"Key Manager - hide for admins if barbican not used","message":"Key Manager - hide for admins if barbican not used\n\nChange-Id: I746c13b80dc6fbd6e795dfc712a5fcea80c0decb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5b116f92e03c81456035ddcc42eebc78320dc147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5b116f92e03c81456035ddcc42eebc78320dc147"}]},"branch":"refs/heads/master","description":"Edit commit message"},"32aa56dab233d7a42d2324ad56797a8c4f2e6ab9":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-12-31 15:44:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/972028/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/972028/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/972028/5"}}},"commit":{"parents":[{"commit":"a51e763a4262b2e254504ea2633bf2319e6590e9","subject":"Merge \"docs: rename field “Available Zone” to “Availability Zone”\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a51e763a4262b2e254504ea2633bf2319e6590e9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:39:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-31 15:44:18.000000000","tz":0},"subject":"Key Manager - hide for admin if barbican not used","message":"Key Manager - hide for admin if barbican not used\n\nChange-Id: I746c13b80dc6fbd6e795dfc712a5fcea80c0decb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32aa56dab233d7a42d2324ad56797a8c4f2e6ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32aa56dab233d7a42d2324ad56797a8c4f2e6ab9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
