)]}'
{"id":"openstack%2Fskyline-console~972996","triplet_id":"openstack%2Fskyline-console~master~If258978084d39bf9dc15137dadd99f521e06c445","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-12 05:08:24.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-12 02:52:58.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-11 19:04:38.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"If258978084d39bf9dc15137dadd99f521e06c445","subject":"Correct various grammar or logic mistakes ","status":"MERGED","created":"2026-01-10 10:31:10.000000000","updated":"2026-01-12 05:09:14.000000000","submitted":"2026-01-12 05:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":72,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972996","meta_rev_id":"6c3b927f27eae1fe0f1f89a6b124976fae1d971a","_number":972996,"virtual_id_number":972996,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 05:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-12 02:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-11 19:04:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-12 02:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-10 11:59:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-10 19:25:47.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-10 19:26:05.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"ebfae7704d2970d45f8cf80bbef03e095f83917a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-10 10:31:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c160cbf436fc5b04ae07416869fcbb7b54c5b68d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 11:59:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e079a27bcbf4a4faf9cc70d8f4cee3a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/27d7725fb7ca4be484482838960c78b9 : SUCCESS in 1m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2af2e01e909049bdb02d40e94b13fd34 : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fa506bbe4346278f3dd6b336c00492 : SUCCESS in 4m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4b38b703e192409b81302f43c81cb171 : SUCCESS in 11m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2b7171db0dde4072bea985f5b148d835 : SUCCESS in 7m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e8d04d5e646440a7be33c88e080b3dec : SUCCESS in 5m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ca333e758c554e899a0363c76cfaec7f : SUCCESS in 1h 23m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7304c87090a44fbb84733d550eba2b98 : SUCCESS in 30m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0006b56289fd4fcba04a060ec6203c8b : SUCCESS in 51m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ade542b33a4a42468ba874bf9d7bc28c : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":1},{"id":"38b88bc26d9b1efeed1747a818ea1621b6d065c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-10 19:14:40.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f2fdf81a605734b3badcf32014234fd2cb5ec920","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-10 19:14:43.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"8ecd15f1222cd4dd1a610beccd9bbd10c4a1042e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 19:30:36.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a3d5a1121fcf4a80e591c8ac5dbe177acb0cf7b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:25:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(17 comments)","accounts_in_message":[],"_revision_number":2},{"id":"49d0145a5130f0e8c814f68a5eff476bc3da8d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 20:26:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/247a4aa929be402282637f47948bbcb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7744450ba8274dccb33073adc062d92f : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/be2a85d030304ff798addf83fcde7f67 : SUCCESS in 11m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e92924a4049e4958b1e749f5113a0fbd : SUCCESS in 5m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/56b9081b9f7a4c3ab16a92a4bbfef359 : SUCCESS in 7m 25s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1754093eba4041de98ae5e5817e810f8 : SUCCESS in 1h 04m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bf6a1acfdd0048a4a8dd3d6e05464839 : FAILURE in 1h 02m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/be413b5ecbc3473da3b762758a14a379 : FAILURE in 58m 20s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c4a7f72621d14dd7ba4799e336a548f7 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":2},{"id":"ca945af72c0dec83dfaa22024e83f3e6fab9c4c3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:51:40.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fa2ac8d3995e1cc35b35a3bd036c7dfc0111750d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:58:26.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0a8740219c74cf03ae5cdd7c39e6ad13e188da7e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:11:05.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"040461189d9e0f4381b00a1d22d70cdd959e607c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:13:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5263520daaa432f083381144b22aae28cc276f12","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:29:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"699617a24033d9ebba10dc5397c17f71e0cb081f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 12:39:03.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b59ee3871373e323f7d7390eb6671e58179caddb","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 12:39:17.000000000","message":"Patch Set 3:\n\n(34 comments)","accounts_in_message":[],"_revision_number":3},{"id":"91fdff60a0275a93a8af059768918e9b1ec15d6f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 12:49:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d24c9c17170dc44ab77e7fd027edf321cfeffebd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 13:41:30.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8919da589d2a54b2abf6db18254c9701c56db31b","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 13:41:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eae8de4b639ee5c433443508240b736642ee2761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 14:48:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19be28248fe34ad6aa234ebf1deb983a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99567fae98b74074a44748e6eece66a5 : SUCCESS in 6m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/83f9069b4dc94907808bbf2e11e63aea : SUCCESS in 10m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f32b34263ac84c0eb8c4312d7d8b0461 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a2e012404f7b44b6a27318397811809c : SUCCESS in 6m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/868124531de640218c87551ab3f5cc4a : SUCCESS in 1h 05m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ef5b2b6daf9c4e4189b59b83fda69057 : FAILURE in 49m 13s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/db9fd3ac855c4470a86fe7d99f7f1af3 : FAILURE in 58m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3b2efde4525c42b5b74e6ef243aa650e : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":4},{"id":"652607faed6ad31c45f02d237dcad0dbc13d5832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 15:41:05.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"70ce3304a5b5e19e3628c2a283716cb501645a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 16:53:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2c8c9f12a5440d482b89d3336d2109f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad168ca3a8f42e2a57ed79a3f672785 : SUCCESS in 4m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4be4dd79209549d294834703c0d62902 : SUCCESS in 6m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/77afcb470fb74b8abd607daed039c1c2 : SUCCESS in 8m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/df59c84df1434145b3706e3f2b9baa0a : SUCCESS in 3m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fd75392ef5204795841d61b4a35d2968 : SUCCESS in 1h 03m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/67a72cd3afc74be1878036f728980b80 : SUCCESS in 59m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/341535c4117e404cbc3d11376ea953a2 : FAILURE in 39m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4019b3325a3f4b668f9f976ee26e2ac8 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":5},{"id":"e808381b9a43d5038fcc32576235ec5b248fb0f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 17:17:17.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5465feefe369dfe69ef00b6075945be36c3f79d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 18:44:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae75d1b5ba504b94874689e21c55f801\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c0ec71fc864981b0d6a5063aa984e1 : SUCCESS in 4m 37s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/22a1920024f84c2ab38e7fa9e2837a6a : SUCCESS in 6m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3ad8dfb87bec4ef484f73f612bfed27a : SUCCESS in 4m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/abe48c59aa624943a73a2e90cff702c2 : SUCCESS in 5m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/746d09f275e5490c902ddb172507f66e : SUCCESS in 1h 22m 03s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7a33e0433efd42f5ac91fd1d042fb5b9 : SUCCESS in 40m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5465b44732aa407689896edae1416e64 : SUCCESS in 42m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8fd57f6da6ae4ffb901db56318442efc : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":6},{"id":"a0622ed95d9e4df14f4952ccac792eb2e972d4fb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 19:04:38.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6b037daac0e5d2e154cca4e9a47e283611c28283","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-12 02:52:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"08bdb6b61121604ed6ebfe96e3e39fea3801ceb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 02:53:30.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4f2468e3360eeedba13b5a2a10f20be1a1c382c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 05:08:24.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a81edc05716b495082a8e0493f1baaf6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f8f2f543f242b4b4a55c4d1b01cdeb : SUCCESS in 6m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9cc7f4626bdc464bac3677640cdcb9a4 : SUCCESS in 7m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0833c384525e4231992606eb96aee32d : SUCCESS in 6m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c8a5a33e98084caba5c2494d199399f4 : SUCCESS in 4m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ff32ff2033f64f87a8232c44119b03b7 : SUCCESS in 1h 21m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e9d0e775eff94f05a10ecda502d51d8a : SUCCESS in 45m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/67753196d6f74bad81b10738d005804b : SUCCESS in 35m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4806e592ebb64274a673a5ba600805a0 : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":6},{"id":"27afc88e9138e902d5875f52a1d7067efb2e04fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 05:08:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6c3b927f27eae1fe0f1f89a6b124976fae1d971a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 05:09:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4e13b53c13a44d5a4161020021f537d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/190ae60731e741f0a34d56b11be0b4d2 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3544e973f9f3c7aec8fd42b005161fc6ff5eedb0","revisions":{"b2ae766795b6070265246a41ed57329b93bb3c97":{"kind":"REWORK","_number":1,"created":"2026-01-10 10:31:10.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/96/972996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/96/972996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/1"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-10 10:31:10.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-10 10:31:10.000000000","tz":0},"subject":"Correct various grammar or logic mistakes ","message":"Correct various grammar or logic mistakes \n\nChange-Id: If258978084d39bf9dc15137dadd99f521e06c445\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b2ae766795b6070265246a41ed57329b93bb3c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b2ae766795b6070265246a41ed57329b93bb3c97"}]},"branch":"refs/heads/master"},"af1d513376bf586520626af07a13062795dd82e8":{"kind":"REWORK","_number":2,"created":"2026-01-10 19:14:40.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/96/972996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/96/972996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/2"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-10 10:31:10.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-10 19:12:32.000000000","tz":0},"subject":"Correct various grammar or logic mistakes ","message":"Correct various grammar or logic mistakes \n\nChange-Id: If258978084d39bf9dc15137dadd99f521e06c445\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af1d513376bf586520626af07a13062795dd82e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af1d513376bf586520626af07a13062795dd82e8"}]},"branch":"refs/heads/master"},"5054ad265be909b522d545c50ca958d2497b6bfd":{"kind":"REWORK","_number":3,"created":"2026-01-11 12:39:03.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/96/972996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/96/972996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/3"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-10 10:31:10.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-11 12:37:56.000000000","tz":0},"subject":"Correct various grammar or logic mistakes ","message":"Correct various grammar or logic mistakes \n\nChange-Id: If258978084d39bf9dc15137dadd99f521e06c445\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5054ad265be909b522d545c50ca958d2497b6bfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5054ad265be909b522d545c50ca958d2497b6bfd"}]},"branch":"refs/heads/master"},"98ef4f7440040213f3b64170e7db0a9b327bc553":{"kind":"REWORK","_number":4,"created":"2026-01-11 13:41:30.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/96/972996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/96/972996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/4"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-10 10:31:10.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-11 13:41:27.000000000","tz":0},"subject":"Correct various grammar or logic mistakes ","message":"Correct various grammar or logic mistakes \n\nChange-Id: If258978084d39bf9dc15137dadd99f521e06c445\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/98ef4f7440040213f3b64170e7db0a9b327bc553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/98ef4f7440040213f3b64170e7db0a9b327bc553"}]},"branch":"refs/heads/master"},"37a51d4e37d23feaa60b8a8a1c0a588ec1744b2c":{"kind":"REWORK","_number":5,"created":"2026-01-11 15:41:05.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/96/972996/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/96/972996/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/5"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-10 10:31:10.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-11 15:41:03.000000000","tz":0},"subject":"Correct various grammar or logic mistakes ","message":"Correct various grammar or logic mistakes \n\nChange-Id: If258978084d39bf9dc15137dadd99f521e06c445\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/37a51d4e37d23feaa60b8a8a1c0a588ec1744b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/37a51d4e37d23feaa60b8a8a1c0a588ec1744b2c"}]},"branch":"refs/heads/master"},"3544e973f9f3c7aec8fd42b005161fc6ff5eedb0":{"kind":"REWORK","_number":6,"created":"2026-01-11 17:17:17.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/96/972996/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/96/972996/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/96/972996/6"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-10 10:31:10.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-11 16:58:00.000000000","tz":0},"subject":"Correct various grammar or logic mistakes ","message":"Correct various grammar or logic mistakes \n\nChange-Id: If258978084d39bf9dc15137dadd99f521e06c445\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3544e973f9f3c7aec8fd42b005161fc6ff5eedb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3544e973f9f3c7aec8fd42b005161fc6ff5eedb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
