)]}'
{"id":"openstack%2Fskyline-console~973007","triplet_id":"openstack%2Fskyline-console~master~I970ebaaf61640acef95776b7c8958e9464dfad41","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-12 08:02:19.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-13 03:56:56.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-13 04:47:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I970ebaaf61640acef95776b7c8958e9464dfad41","subject":"VCPU, VGPU replaced to vCPU, vGPU","status":"MERGED","created":"2026-01-10 20:29:37.000000000","updated":"2026-01-13 04:48:42.000000000","submitted":"2026-01-13 04:47:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973007","meta_rev_id":"8f5442eebeba691291ad8913d639f9a8dd633519","_number":973007,"virtual_id_number":973007,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-13 04:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-12 08:59:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-13 03:56:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-13 03:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-10 20:33:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-01-10 20:41:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-10 23:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-11 13:12:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"25cd732f3a225066d0830566413e78279c169daf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:29:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9baae84428a414b20e201372bacf158fea82d639","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:31:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a0d075f62810defd6523c1ce48bbe8b2c283c612","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:32:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a08eaafd9e3af79c829688ee0a58da3b543b3b5e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:33:05.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"169766e004a1ee2d20ac7b2618ab4bbc3c9dcd58","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:34:21.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a28849301e6836c2c6a3f167ab6bf678a0a3422d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:34:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6f36eae65bb9098afdbe36a52354912372eae250","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:35:27.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"af4ac4943b9fd988164da1352970e451bc485c54","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:39:14.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"b335630f66b9010117dd3a048ab40d6edfbd284c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:41:02.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"b7115e9085f0d689d1775ef6ffdb74759e86b9c0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:41:15.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":7},{"id":"240a58bdb2fe2f0b93abe4740307ffd02531fc6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:42:18.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2132eb7eff0d04dff00c34aae36e200049ab29c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:44:29.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c76895eab628c14cae2df5bd856a45919b039aea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 20:46:15.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"a136dbf352017ed80a87077150ab0718dc448cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:20:58.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"018a6b3c12701b04ce18aee540e1716e5deb8c0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:21:17.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"29a9ed0f7701c722bb096777ecedc3b93114f06c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:26:50.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"9076179e4673bb3046ea667962780284a27da1a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:30:59.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"265c13e569c00aec75be9d946fcdc86dcc08085c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:37:07.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"d16a548bb3bd367e453ba2d1ba3a7593cd516937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:38:28.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"9b115aa63f00f557a64ea026344681c8ad33083b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:42:55.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"d486060b63c9f853141e5f1f871fe8102278472c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:43:24.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"61adbc9e87f0798eade785e3dd4354bbbce6493c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:44:37.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"e91c175da61297eb3edef1d652949085f50f17c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:45:52.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"97778cad314f73be37fd13f97babc802fa0b6617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:48:01.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"dc78684b971dfa6d75885ebc7fd98f268336c97b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:49:15.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"dcd4ed348685f282e5ab508fb776566e24c8d0c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:51:02.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"5d3973aa4c7f9ee8b235e80024ce87a2095109c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:51:35.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"e31a65ba03442bfd6b959c6f0fefbb69348d615e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:53:05.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"3b29f8080a9f7fee917461a2ef4e6827bd60a216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:54:55.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"e9e75bbaf5884e615c929863dbb175c0e1c6687c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:56:19.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"bd01a776e9c2de9d1848946f2e8d65aa77010363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 21:59:25.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"54a70e4283f912f6ba44263484742ca132baa371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 22:00:33.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"8f6533dec1f3a8be70e97c26edb128dcc5be7e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 22:00:59.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"4bfd088c3fe5744e9c9087e005ce35e62d9b7458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-10 22:01:08.000000000","message":"Patch Set 30: Commit message was updated.","accounts_in_message":[],"_revision_number":30},{"id":"0526f688d6e801e4b42c1e3ed0b1c2f180f8ee33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 23:28:13.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b91ad80c916145069880e7449dc4a006\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b3e774589c646698bdfc0d215644d5c : SUCCESS in 7m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1d869dded6a649f1b2c0eb694c7a9d44 : SUCCESS in 10m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/29b5a364dbb642a98f0b6f6bfff5d31d : SUCCESS in 6m 01s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fbec66b845a244c0aeeddeca3df8ebff : SUCCESS in 4m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/37457383f6024262ba059b63fffcd9d5 : SUCCESS in 1h 21m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/56e3c4e2798847cebd60953a224d53a6 : SUCCESS in 1h 00m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f5e568f6d4db457aaaf4644d4a73dad8 : SUCCESS in 52m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9d023bb74f174eccb7cbf28f3ea15198 : FAILURE in 53m 01s","accounts_in_message":[],"_revision_number":30},{"id":"e9a1a8cfede0a02afbe8aa62243b3570ef428014","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:27:43.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"91de09c71d5f505f42cbe048312f27aa4b7c5c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:35:32.000000000","message":"Patch Set 31: Published edit on patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"f13a14ea5a38956f237cfc66553c818baf463dda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:37:43.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"2f8c605852f8cf83d0d869a9f998c634059abfd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:43:27.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"5dbc38e837f82d4cdf5062a8d6da54702e085c3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:46:47.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"534c0db36810e201a43db8a82f55e563263d7ef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:48:49.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"aaa64363f61c0aa18e3051770b501de0570d98c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:51:25.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"39d08f01ca73aeff2b963a2e5fa56a3a8f25e9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:51:51.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"6593d43204c0ae12d7246710a554fb04e8587651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:53:20.000000000","message":"Patch Set 38: Published edit on patch set 37.","accounts_in_message":[],"_revision_number":38},{"id":"44286406d8506ce85864bef5db64b706124d6846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:56:02.000000000","message":"Patch Set 39: Published edit on patch set 38.","accounts_in_message":[],"_revision_number":39},{"id":"0f2aa502375985951b3253db52843a36a3687ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:56:25.000000000","message":"Patch Set 40: Published edit on patch set 39.","accounts_in_message":[],"_revision_number":40},{"id":"ec156a5dd22c4ac974d590075d925b045ae5fce7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 10:56:59.000000000","message":"Patch Set 41: Published edit on patch set 40.","accounts_in_message":[],"_revision_number":41},{"id":"950a8d012f077d0b09becc85549411993bab3b32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 11:02:47.000000000","message":"Patch Set 42: Published edit on patch set 41.","accounts_in_message":[],"_revision_number":42},{"id":"e46137d0a0a44935bb86091f9a864c8acb5c9613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 12:07:15.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/371468c6a6dd44fc9e49bddfb19fbe72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbceb6b3b6d4764aee99f2cdf971b45 : SUCCESS in 3m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4de1696be03a43d2a50d12de69575b80 : SUCCESS in 6m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f864febe73804d2693653852e85a535f : SUCCESS in 6m 35s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/66c0a07288fa49aaad6f098653c6040b : SUCCESS in 6m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7095de18bbc54d20bf0a38f24eb6cbd5 : SUCCESS in 58m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3e10733d237c4d04b8adc01235eb78d9 : SUCCESS in 59m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7ac22de61ece409b8948b594c1d9281d : SUCCESS in 36m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5dc0070838784425b3793836ac5076cc : FAILURE in 53m 16s","accounts_in_message":[],"_revision_number":42},{"id":"4a0a56cc41e8ce0f55dbbf17cd5f0abdf24193ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 12:18:10.000000000","message":"Patch Set 43: Published edit on patch set 42.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":43},{"id":"de05ff379fc1e1c2ba3ab5be4a4895e67cfa7fb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 12:43:05.000000000","message":"Patch Set 44: Published edit on patch set 43.","accounts_in_message":[],"_revision_number":44},{"id":"35738fface875483afc2e4a7f2513bb0593e9ca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 12:43:55.000000000","message":"Patch Set 45: Published edit on patch set 44.","accounts_in_message":[],"_revision_number":45},{"id":"d273c59e7f179ecb0561be58d534341a080f4426","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 13:08:32.000000000","message":"Patch Set 43:\n\n(3 comments)","accounts_in_message":[],"_revision_number":43},{"id":"3142f8970ef08259b412b80fa1908599e19ab73c","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 13:10:21.000000000","message":"Patch Set 45:\n\n(2 comments)","accounts_in_message":[],"_revision_number":45},{"id":"f759fd7b7f4b4b399e5e1a4b72c4fed54039cb72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 13:11:34.000000000","message":"Patch Set 46: Published edit on patch set 45.","accounts_in_message":[],"_revision_number":46},{"id":"35aef7252ffdc466b5b6659199477591582cdcd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 13:12:08.000000000","message":"Patch Set 47: Published edit on patch set 46.","accounts_in_message":[],"_revision_number":47},{"id":"0275e7b0b7da27646982f36e9b3bd5efa7db9e2e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 13:12:25.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"58920aef371188190e81882918a484f3ec6b8398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 14:36:56.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e8e867f755491b864796edb169cd08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a1562d6ecc4689b90f01cf9c57e756 : SUCCESS in 6m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/54e4458822b74c97add7712784c2915a : SUCCESS in 11m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/67e9ef4e3b2c4937bcaa4ebaa56192ee : SUCCESS in 8m 15s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/873f323a8a124c24b26345c0bff6e8ab : SUCCESS in 7m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ab056f32db0a4d95a8cb1402dbe0958b : SUCCESS in 1h 20m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ec47875538454218b5afe059306d212a : SUCCESS in 58m 52s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f07ecd6897e04b939f02e36cf2cc019b : SUCCESS in 39m 26s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/449049c4892c4027ae620c43713a7e3b : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":47},{"id":"e421c60a7b03ff91af9064d9d4dffd318a0c6031","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-12 02:49:45.000000000","message":"Patch Set 47: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":47},{"id":"71e93e15bfe5a8d0a820555fa9f078b873b3f2cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 04:45:41.000000000","message":"Patch Set 48: Published edit on patch set 47.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":48},{"id":"8630711658a2f1cdf0f103e5c85c164e217d1676","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 04:45:49.000000000","message":"Patch Set 48:\n\n(2 comments)","accounts_in_message":[],"_revision_number":48},{"id":"b0489008e828c4953a9a2ee07dac726a8188c970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 06:12:11.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4394f1f15ce14e82929200ebc3f3f5e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f39c6397f764d0c97c83effbe4d9401 : SUCCESS in 3m 42s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6f0a153c583f45baad9882447f6da8d5 : SUCCESS in 6m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/db214b43963d46c7b75780f8e3012106 : SUCCESS in 3m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cf5f0a00f6664ea4ab43185aa3c916eb : SUCCESS in 3m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8972c981ba6c4e81b1c207120a16d2f2 : SUCCESS in 1h 21m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/095c569c35ac47859a04a21135ce1137 : SUCCESS in 1h 01m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/76912c06e560428fa63d1af4d2c27af7 : SUCCESS in 49m 35s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e7f9568d7f4e4f15bc31c6c7a50b32ce : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":48},{"id":"7f2b19c17ef4f0495f09f8fe32fa3297d99eb19b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 06:21:01.000000000","message":"Patch Set 49: Published edit on patch set 48.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":49},{"id":"b6c68f472f8ee5d7011854ec7c59226f286b134d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:41:55.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c0b4891fdac4f139b1098dee5860909\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb09af2a37ba41a0bcdcad7142bf4280 : SUCCESS in 5m 55s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a138c61153ac4cddb4c01cdf1375ec20 : SUCCESS in 5m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d3dfe8de53fd4e50b576f10055480cf8 : SUCCESS in 7m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3c2dc348e8d2462ba12d21983b3c2c0a : SUCCESS in 4m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e048c8be26ea4923b71de37116226780 : SUCCESS in 1h 19m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/62723b6e2b8149a8a207b325df3df5ff : FAILURE in 19m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fba31a4ccb504f71b5b8a4766b71fee4 : SUCCESS in 50m 20s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0ddab5694bfc4c799975a6e86b2da773 : SUCCESS in 36m 44s","accounts_in_message":[],"_revision_number":49},{"id":"c0ebb3f3c416a8643e9ebb20acdfd3ca15a2a433","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-12 08:02:19.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"a0af27f426022a47a1f4abe6fdc005ce7afa81ff","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-12 08:59:52.000000000","message":"Patch Set 49: Code-Review+1","accounts_in_message":[],"_revision_number":49},{"id":"d0bf59ea84538f081138697931c6050b78b2a955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 09:04:03.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abbfdbb29b104028a268997f5eb2be46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff619a835c741349c49c312bf9fbd06 : SUCCESS in 3m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/718130f9c9434996a2c2557feaaae5c5 : SUCCESS in 6m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0b55e4061f6449d1bde12dd2549e2263 : SUCCESS in 7m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3efe532a994c485d85304d8a0637ba08 : SUCCESS in 4m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/beacf8128dfd4c83887afda20de039b0 : SUCCESS in 50m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/689a29491fbc43fc8391bdbb674bac5a : SUCCESS in 1h 00m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/62fda3a3f6744a388ec21456fdf450a6 : SUCCESS in 53m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/aca88468517e422b9f007328374500d8 : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":49},{"id":"1ec958e05b131a6262f8a1456a9914d92cd05c73","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-13 03:56:56.000000000","message":"Patch Set 49: Code-Review+2","accounts_in_message":[],"_revision_number":49},{"id":"820a25b8164617cfe6918d56255a8af4bebb115d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-13 03:57:02.000000000","message":"Patch Set 49: Workflow+1","accounts_in_message":[],"_revision_number":49},{"id":"8b5851beacb80515974fd5640747ef250f669bc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 03:57:32.000000000","message":"Patch Set 49: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":49},{"id":"5576a1238c3a6ad696dbd1d23ad1a734c6df4276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 04:47:52.000000000","message":"Patch Set 49: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ab19bc1a2d46bca8d1d53337ff9dd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e36f5445a6f454a9985fec7bbc05732 : SUCCESS in 4m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b2fe00c53b4b4cf585fbfbf349a3a0aa : SUCCESS in 5m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0042ec3abd3041c0b269bd70097d380d : SUCCESS in 3m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/901cc42204744fb3be193c005874f33f : SUCCESS in 3m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ba669f06bd6f4ca1973bca83d15fd8cc : SUCCESS in 49m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a349543957f24c56b3f077d8254f4372 : SUCCESS in 33m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9b4aac249bb8467787eb75096217950a : SUCCESS in 30m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/76c860d1fca54a42adbb82d33d8b846e : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":49},{"id":"580ffe2c55aea85813d40747da51dffe860e786e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 04:47:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":49},{"id":"8f5442eebeba691291ad8913d639f9a8dd633519","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 04:48:42.000000000","message":"Patch Set 49:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7245e063c2c1420c933c8bf43d693e28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e827c2ae5f14ed68405e77a6f4cb070 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":49}],"current_revision_number":49,"current_revision":"9d0a82a8c58afe61521fbb0eca9388a0c6bfbcaa","revisions":{"c931fe99ac008c90fa23300f0506a185ac55c455":{"kind":"REWORK","_number":1,"created":"2026-01-10 20:29:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/1"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c931fe99ac008c90fa23300f0506a185ac55c455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c931fe99ac008c90fa23300f0506a185ac55c455"}]},"branch":"refs/heads/master"},"dfb48f0de2b0201c31ab774648f4689b74379d7f":{"kind":"REWORK","_number":2,"created":"2026-01-10 20:31:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/2"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:31:00.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dfb48f0de2b0201c31ab774648f4689b74379d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dfb48f0de2b0201c31ab774648f4689b74379d7f"}]},"branch":"refs/heads/master"},"ad8f966ede64b282972d213218946cf233a86620":{"kind":"REWORK","_number":3,"created":"2026-01-10 20:32:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/3"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:32:10.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ad8f966ede64b282972d213218946cf233a86620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ad8f966ede64b282972d213218946cf233a86620"}]},"branch":"refs/heads/master"},"7b593e8eea8d9afa71f176aa32bb513128fe24b5":{"kind":"REWORK","_number":4,"created":"2026-01-10 20:34:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/4"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:34:21.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b593e8eea8d9afa71f176aa32bb513128fe24b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b593e8eea8d9afa71f176aa32bb513128fe24b5"}]},"branch":"refs/heads/master"},"67cbe831e5cca7425a314513e33957e779bdc94b":{"kind":"REWORK","_number":5,"created":"2026-01-10 20:35:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/5"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:35:27.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/67cbe831e5cca7425a314513e33957e779bdc94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/67cbe831e5cca7425a314513e33957e779bdc94b"}]},"branch":"refs/heads/master"},"b778ef7110ec8ef41926d2c3f44c2ff513d30d03":{"kind":"REWORK","_number":6,"created":"2026-01-10 20:39:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/6"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:39:14.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b778ef7110ec8ef41926d2c3f44c2ff513d30d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b778ef7110ec8ef41926d2c3f44c2ff513d30d03"}]},"branch":"refs/heads/master"},"d4e1f9cc119b42a8abeb58c3bdbe250f978c0287":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-01-10 20:41:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/7"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:41:02.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d4e1f9cc119b42a8abeb58c3bdbe250f978c0287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d4e1f9cc119b42a8abeb58c3bdbe250f978c0287"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c979540333956ae4e76b305d8fd98aeb3494e863":{"kind":"REWORK","_number":8,"created":"2026-01-10 20:42:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/8"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:42:18.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c979540333956ae4e76b305d8fd98aeb3494e863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c979540333956ae4e76b305d8fd98aeb3494e863"}]},"branch":"refs/heads/master"},"c4816e15efcf9af9951814c52df2c6dc0a9bd876":{"kind":"REWORK","_number":9,"created":"2026-01-10 20:44:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/9"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:44:29.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c4816e15efcf9af9951814c52df2c6dc0a9bd876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c4816e15efcf9af9951814c52df2c6dc0a9bd876"}]},"branch":"refs/heads/master"},"2fecd3794fd7e31429687e01e73a0d5fc795826d":{"kind":"REWORK","_number":10,"created":"2026-01-10 20:46:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/10"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:46:15.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2fecd3794fd7e31429687e01e73a0d5fc795826d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2fecd3794fd7e31429687e01e73a0d5fc795826d"}]},"branch":"refs/heads/master"},"96f21c15e111a78085c6326e1637023cb3003582":{"kind":"REWORK","_number":11,"created":"2026-01-10 21:20:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/11"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:20:58.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/96f21c15e111a78085c6326e1637023cb3003582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/96f21c15e111a78085c6326e1637023cb3003582"}]},"branch":"refs/heads/master"},"5e6eb75d8270eff86d9e6a8b6caea3474a983a08":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2026-01-10 21:21:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/12"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:21:17.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5e6eb75d8270eff86d9e6a8b6caea3474a983a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5e6eb75d8270eff86d9e6a8b6caea3474a983a08"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8bf47a10ee3674d98ef15eafb10f3961c0946380":{"kind":"REWORK","_number":13,"created":"2026-01-10 21:26:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/13"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:26:50.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8bf47a10ee3674d98ef15eafb10f3961c0946380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8bf47a10ee3674d98ef15eafb10f3961c0946380"}]},"branch":"refs/heads/master"},"1f7011af843d73272d0da657ea51ad8feb9f5341":{"kind":"REWORK","_number":14,"created":"2026-01-10 21:30:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/14"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:30:59.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1f7011af843d73272d0da657ea51ad8feb9f5341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1f7011af843d73272d0da657ea51ad8feb9f5341"}]},"branch":"refs/heads/master"},"2818166fe2fa7066f67ddac7fff9ff78d1f8495e":{"kind":"REWORK","_number":15,"created":"2026-01-10 21:37:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/15"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:37:06.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2818166fe2fa7066f67ddac7fff9ff78d1f8495e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2818166fe2fa7066f67ddac7fff9ff78d1f8495e"}]},"branch":"refs/heads/master"},"c248366e8de793f3e444eec5cbde1370d7c904fe":{"kind":"REWORK","_number":16,"created":"2026-01-10 21:38:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/16"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:38:28.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c248366e8de793f3e444eec5cbde1370d7c904fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c248366e8de793f3e444eec5cbde1370d7c904fe"}]},"branch":"refs/heads/master"},"d2c1b3913729fff8d05b228b163ab408763d90ce":{"kind":"REWORK","_number":17,"created":"2026-01-10 21:42:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/17"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:42:54.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d2c1b3913729fff8d05b228b163ab408763d90ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d2c1b3913729fff8d05b228b163ab408763d90ce"}]},"branch":"refs/heads/master"},"7ab5d7faa257c3ac9faf51178152a77ac010cf9d":{"kind":"REWORK","_number":18,"created":"2026-01-10 21:44:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/18"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:44:36.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ab5d7faa257c3ac9faf51178152a77ac010cf9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ab5d7faa257c3ac9faf51178152a77ac010cf9d"}]},"branch":"refs/heads/master"},"62d5bd400d17e6fba8322ea2783d6fbfa52017aa":{"kind":"REWORK","_number":19,"created":"2026-01-10 21:45:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/19"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:45:51.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/62d5bd400d17e6fba8322ea2783d6fbfa52017aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/62d5bd400d17e6fba8322ea2783d6fbfa52017aa"}]},"branch":"refs/heads/master"},"d85b9dc213a103c2bd83db799f3d2a500ffb46e8":{"kind":"REWORK","_number":20,"created":"2026-01-10 21:48:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/20"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:48:01.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d85b9dc213a103c2bd83db799f3d2a500ffb46e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d85b9dc213a103c2bd83db799f3d2a500ffb46e8"}]},"branch":"refs/heads/master"},"0ac70242b7d465f1de55d6f758304036a4b4ab24":{"kind":"REWORK","_number":21,"created":"2026-01-10 21:49:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/21"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:49:10.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0ac70242b7d465f1de55d6f758304036a4b4ab24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0ac70242b7d465f1de55d6f758304036a4b4ab24"}]},"branch":"refs/heads/master"},"f4188cc24503e929947857aa20aa94ffe526fa39":{"kind":"REWORK","_number":22,"created":"2026-01-10 21:51:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/22"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:51:02.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4188cc24503e929947857aa20aa94ffe526fa39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4188cc24503e929947857aa20aa94ffe526fa39"}]},"branch":"refs/heads/master"},"752e1db5a765e0f29b4b1c465ee134e6075ade3b":{"kind":"REWORK","_number":23,"created":"2026-01-10 21:51:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/23"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:51:35.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/752e1db5a765e0f29b4b1c465ee134e6075ade3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/752e1db5a765e0f29b4b1c465ee134e6075ade3b"}]},"branch":"refs/heads/master"},"de37647d3828c4c7126514cef8c51ce9a1c01488":{"kind":"REWORK","_number":24,"created":"2026-01-10 21:53:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/24"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:53:05.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/de37647d3828c4c7126514cef8c51ce9a1c01488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/de37647d3828c4c7126514cef8c51ce9a1c01488"}]},"branch":"refs/heads/master"},"d417334d40787fab27d8ac9f52a022fd10f8537b":{"kind":"REWORK","_number":25,"created":"2026-01-10 21:54:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/25"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:54:55.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d417334d40787fab27d8ac9f52a022fd10f8537b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d417334d40787fab27d8ac9f52a022fd10f8537b"}]},"branch":"refs/heads/master"},"6d4cbae6c36cc2003bdbc8571a8c395d425b2be9":{"kind":"REWORK","_number":26,"created":"2026-01-10 21:56:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/26"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:56:19.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6d4cbae6c36cc2003bdbc8571a8c395d425b2be9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6d4cbae6c36cc2003bdbc8571a8c395d425b2be9"}]},"branch":"refs/heads/master"},"6749374d74794efa9e114a00239de4c7f3310081":{"kind":"REWORK","_number":27,"created":"2026-01-10 21:59:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/27"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 21:59:25.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6749374d74794efa9e114a00239de4c7f3310081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6749374d74794efa9e114a00239de4c7f3310081"}]},"branch":"refs/heads/master"},"0ecf194ed674e0d1c38e2b579930ba7e237be1f5":{"kind":"REWORK","_number":28,"created":"2026-01-10 22:00:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/28"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 22:00:32.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0ecf194ed674e0d1c38e2b579930ba7e237be1f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0ecf194ed674e0d1c38e2b579930ba7e237be1f5"}]},"branch":"refs/heads/master"},"31ac16b97a45dcc38449c3dacf1dae4fe4c60bf9":{"kind":"REWORK","_number":29,"created":"2026-01-10 22:00:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/29"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 22:00:59.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes for networks.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/31ac16b97a45dcc38449c3dacf1dae4fe4c60bf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/31ac16b97a45dcc38449c3dacf1dae4fe4c60bf9"}]},"branch":"refs/heads/master"},"c7bf13ddbfc7dde6c06f5c2114ffa9da8cc49dc6":{"kind":"NO_CODE_CHANGE","_number":30,"created":"2026-01-10 22:01:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/30"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 22:01:08.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c7bf13ddbfc7dde6c06f5c2114ffa9da8cc49dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c7bf13ddbfc7dde6c06f5c2114ffa9da8cc49dc6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2a6bf3004428893dea261177be798e5a6112ed9b":{"kind":"REWORK","_number":31,"created":"2026-01-11 10:35:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/31"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:35:32.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2a6bf3004428893dea261177be798e5a6112ed9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2a6bf3004428893dea261177be798e5a6112ed9b"}]},"branch":"refs/heads/master"},"5d9c31733b0497c71a76030c320ecf07222e313e":{"kind":"REWORK","_number":32,"created":"2026-01-11 10:37:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/32"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:37:43.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d9c31733b0497c71a76030c320ecf07222e313e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d9c31733b0497c71a76030c320ecf07222e313e"}]},"branch":"refs/heads/master"},"c54c5ea74a8a7cfcb6e00050a70a2a7f662f7c92":{"kind":"REWORK","_number":33,"created":"2026-01-11 10:43:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/33"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:43:27.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c54c5ea74a8a7cfcb6e00050a70a2a7f662f7c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c54c5ea74a8a7cfcb6e00050a70a2a7f662f7c92"}]},"branch":"refs/heads/master"},"e63cfcbf26e5151e1bebe5832ff59f48fc09cc28":{"kind":"REWORK","_number":34,"created":"2026-01-11 10:46:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/34"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:46:46.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e63cfcbf26e5151e1bebe5832ff59f48fc09cc28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e63cfcbf26e5151e1bebe5832ff59f48fc09cc28"}]},"branch":"refs/heads/master"},"0facc2464713fa572632d680cb8f8798c2c5bd29":{"kind":"REWORK","_number":35,"created":"2026-01-11 10:48:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/35"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:48:49.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0facc2464713fa572632d680cb8f8798c2c5bd29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0facc2464713fa572632d680cb8f8798c2c5bd29"}]},"branch":"refs/heads/master"},"b9fac14cc1d2b57e3445dbc8951beada0d870c93":{"kind":"REWORK","_number":36,"created":"2026-01-11 10:51:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/36"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:51:25.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b9fac14cc1d2b57e3445dbc8951beada0d870c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b9fac14cc1d2b57e3445dbc8951beada0d870c93"}]},"branch":"refs/heads/master"},"14ef61f50dabd72013e0f18af0aac5efe6162cbe":{"kind":"REWORK","_number":37,"created":"2026-01-11 10:51:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/37"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:51:50.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/14ef61f50dabd72013e0f18af0aac5efe6162cbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/14ef61f50dabd72013e0f18af0aac5efe6162cbe"}]},"branch":"refs/heads/master"},"4f54705559bfb3152fc76bc636840dbad3559a33":{"kind":"REWORK","_number":38,"created":"2026-01-11 10:53:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/38"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:53:20.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4f54705559bfb3152fc76bc636840dbad3559a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4f54705559bfb3152fc76bc636840dbad3559a33"}]},"branch":"refs/heads/master"},"83ae9302905cfb4f4c87380f77ef432afc71c8b8":{"kind":"REWORK","_number":39,"created":"2026-01-11 10:56:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/39"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:56:02.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/83ae9302905cfb4f4c87380f77ef432afc71c8b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/83ae9302905cfb4f4c87380f77ef432afc71c8b8"}]},"branch":"refs/heads/master"},"9e0a79123fdc142973dde0656ba1071301e40f5e":{"kind":"REWORK","_number":40,"created":"2026-01-11 10:56:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/40"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:56:24.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9e0a79123fdc142973dde0656ba1071301e40f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9e0a79123fdc142973dde0656ba1071301e40f5e"}]},"branch":"refs/heads/master"},"af37fd2634011dfc8a97bc79bc5ba7b65ac8190e":{"kind":"REWORK","_number":41,"created":"2026-01-11 10:56:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/41"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 10:56:58.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af37fd2634011dfc8a97bc79bc5ba7b65ac8190e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af37fd2634011dfc8a97bc79bc5ba7b65ac8190e"}]},"branch":"refs/heads/master"},"a1c500fcef676dad9744f98dc1fa9624edeb3465":{"kind":"REWORK","_number":42,"created":"2026-01-11 11:02:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/42"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 11:02:47.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1c500fcef676dad9744f98dc1fa9624edeb3465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1c500fcef676dad9744f98dc1fa9624edeb3465"}]},"branch":"refs/heads/master"},"a53fa256ef1a46255b81d91663a5433c8e299df7":{"kind":"REWORK","_number":43,"created":"2026-01-11 12:18:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/43"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 12:18:10.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a53fa256ef1a46255b81d91663a5433c8e299df7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a53fa256ef1a46255b81d91663a5433c8e299df7"}]},"branch":"refs/heads/master"},"8819d76f49b8ce4ae93e09566b4de4783e1ed7cb":{"kind":"REWORK","_number":44,"created":"2026-01-11 12:43:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/44"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 12:43:05.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8819d76f49b8ce4ae93e09566b4de4783e1ed7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8819d76f49b8ce4ae93e09566b4de4783e1ed7cb"}]},"branch":"refs/heads/master"},"3e35d42484576b82b54e16cc2648ad64f0acfbb1":{"kind":"REWORK","_number":45,"created":"2026-01-11 12:43:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/45"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 12:43:55.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e35d42484576b82b54e16cc2648ad64f0acfbb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e35d42484576b82b54e16cc2648ad64f0acfbb1"}]},"branch":"refs/heads/master"},"51d63105cd94f6ad3aec3cc5f7b4aaab465f4b23":{"kind":"REWORK","_number":46,"created":"2026-01-11 13:11:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/46"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 13:11:34.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/51d63105cd94f6ad3aec3cc5f7b4aaab465f4b23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/51d63105cd94f6ad3aec3cc5f7b4aaab465f4b23"}]},"branch":"refs/heads/master"},"cdfe6b64d000e9bcc6f042d25ffa8a3a3f29abae":{"kind":"REWORK","_number":47,"created":"2026-01-11 13:12:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/47"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-11 13:12:08.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cdfe6b64d000e9bcc6f042d25ffa8a3a3f29abae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cdfe6b64d000e9bcc6f042d25ffa8a3a3f29abae"}]},"branch":"refs/heads/master"},"6608c54d2879029e06024f9a0e1e0f3e6a2b2122":{"kind":"REWORK","_number":48,"created":"2026-01-12 04:45:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/48"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 04:45:41.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6608c54d2879029e06024f9a0e1e0f3e6a2b2122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6608c54d2879029e06024f9a0e1e0f3e6a2b2122"}]},"branch":"refs/heads/master"},"9d0a82a8c58afe61521fbb0eca9388a0c6bfbcaa":{"kind":"REWORK","_number":49,"created":"2026-01-12 06:21:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973007/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973007/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973007/49"}}},"commit":{"parents":[{"commit":"820d6faa51ae38b0d4a1160f4d902cbdce124c1e","subject":"Merge \"Port Range - add possible translation for Any\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/820d6faa51ae38b0d4a1160f4d902cbdce124c1e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-10 20:29:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 06:21:01.000000000","tz":0},"subject":"VCPU, VGPU replaced to vCPU, vGPU","message":"VCPU, VGPU replaced to vCPU, vGPU\n\nWriting style of units of measurement has been updated to be more correct.\n\nAnd few small fixes.\n\nChange-Id: I970ebaaf61640acef95776b7c8958e9464dfad41\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9d0a82a8c58afe61521fbb0eca9388a0c6bfbcaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9d0a82a8c58afe61521fbb0eca9388a0c6bfbcaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
