)]}'
{"id":"openstack%2Fskyline-console~973078","triplet_id":"openstack%2Fskyline-console~stable%2F2024.2~I91a7ccb0a106256a0cbd40291b486c05502d9e05","project":"openstack/skyline-console","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-12 16:09:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I91a7ccb0a106256a0cbd40291b486c05502d9e05","subject":"Monkey patch createHash function as workaround for hardcoded md4 hash","status":"ABANDONED","created":"2026-01-12 12:09:52.000000000","updated":"2026-01-12 16:09:15.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":950459,"cherry_pick_of_patch_set":1,"meta_rev_id":"faf9e209deb78f10a2a9f52bb9930b1b61ffeecd","_number":973078,"virtual_id_number":973078,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-12 12:54:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-12 12:09:52.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"CC"},{"updated":"2026-01-12 12:54:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 13:00:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"25d2606f4d02706e5fe092c1bdf7d707b86b4291","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-12 12:09:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"167d8541c9cf96e2ca6a6e13809876f40d2ba6fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 12:54:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00766f975a2d44c6be319c184e0f5fc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/123e90ad881b40a490afddee4ae9f40e : SUCCESS in 3m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ed0521dc49c14f13b16115364b021d7b : SUCCESS in 9m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bee877f5e5744812b38176a6097b9e3c : SUCCESS in 7m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/59b30f805dc5485e8911fb286f7c23e7 : SUCCESS in 3m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/33bf630771cf4fc99646555be0f56557 : RETRY_LIMIT in 12m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5cd541403e2140ed961fe6ef9dee4ef3 : RETRY_LIMIT in 6m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b14acbb6ae6243bab04d78460f0b425b : RETRY_LIMIT in 10m 03s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/24059371981e4179a03bc74a5441664d : RETRY_LIMIT in 9m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5e11614d18e271059763b00921e09a2e2836181b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 13:00:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"912bd78107d1855fbfc851a6be43721079021a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 13:54:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4d033fa43024b879fa9b88044a5e3dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ec619bfd694f47bbb93f6c614e34fa : SUCCESS in 4m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7b960355ef5449faa7bc0df9b55126dd : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/538e9b3172f44cf69f0d2dbf50be2f40 : SUCCESS in 8m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1084325c9a6c47bba432dc2e9a45b38f : SUCCESS in 3m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/eecb3afac9b84dcba9195ab43d5fecfe : RETRY_LIMIT in 12m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bd10eb5652a2402eb37dbc1dde79423c : RETRY_LIMIT in 10m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/996daa4872944d729eb2d9468f154557 : RETRY_LIMIT in 6m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/50034d595fed4d979558c8be89b846b8 : RETRY_LIMIT in 9m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ce03cc943646a3b577933639bae682168decb78b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 15:44:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"faf9e209deb78f10a2a9f52bb9930b1b61ffeecd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-12 16:09:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e59f07bc63272f164a47ca08dff6da8ffbb9e40","revisions":{"6e59f07bc63272f164a47ca08dff6da8ffbb9e40":{"kind":"REWORK","_number":1,"created":"2026-01-12 12:09:52.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/78/973078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/78/973078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/973078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/973078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/973078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/78/973078/1"}}},"commit":{"parents":[{"commit":"da4d16212f1e6cbfbb2ddde68722f24383a6820c","subject":"Merge \"feat: update package for release\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/da4d16212f1e6cbfbb2ddde68722f24383a6820c"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2025-05-20 19:37:14.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-12 12:09:52.000000000","tz":0},"subject":"Monkey patch createHash function as workaround for hardcoded md4 hash","message":"Monkey patch createHash function as workaround for hardcoded md4 hash\n\nThis patch modifies createHash at runtime to workaround hardcoded\nhashing algorithms (md4) in various Javascript modules that are\nno longer supported by default in OpenSSL included in Ubuntu 24.04\nand possibly other operating systems. Allows for successful\n\"yarn run build\".\n\nChange-Id: I91a7ccb0a106256a0cbd40291b486c05502d9e05\n(cherry picked from commit 580f14adb8a07b11769e7391c24755b5ace1cdc6)\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6e59f07bc63272f164a47ca08dff6da8ffbb9e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6e59f07bc63272f164a47ca08dff6da8ffbb9e40"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
